)]}'
{"id":"openstack%2Fzun~679573","triplet_id":"openstack%2Fzun~master~I00bd54b6d71b3e72499ff9e906f2d0a1cdf4e284","project":"openstack/zun","branch":"master","topic":"bug/1842243","hashtags":[],"change_id":"I00bd54b6d71b3e72499ff9e906f2d0a1cdf4e284","subject":"Exponential backoff on retrying create network","status":"MERGED","created":"2019-09-01 16:39:11.000000000","updated":"2019-09-02 01:37:15.000000000","submitted":"2019-09-02 01:35:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679573-1567388154324-345d384a","meta_rev_id":"5ab7fb1b5b230500ce744fe1fdc805deb27ee536","_number":679573,"virtual_id_number":679573,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"date":"2019-09-01 18:46:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-02 01:35:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-09-02 00:25:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-09-02 00:25:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-01 18:46:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-02 01:35:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c85c44220e7a6b95a84487268ac6fa29c8355c0d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-01 16:39:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ca0dcd8a1b046e2d5b26d288a0624dd5f3819d7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-01 16:39:26.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"602fc4d6aee5a7fe761727a35ed5f894078ef703","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-01 17:21:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c641188a77a8601003c4c9760f2f6475de8e2219","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 18:28:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1560638842564e3eab220d55cdf3992e : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/addfb168f8b14155b0e6843a1a70bf87 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6776a3f8a28649aa9ea078601c8739fd : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85828b0223034491bcd07f96c676202e : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02b3d67148fb4bb69bcfb073d3bac593 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/355c881663414be3b2880236ebcedc72 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a2c704ee3d444c89f5c3ae53cfc2dc : SUCCESS in 4m 55s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/0e18b1c08a844a7e9a6e0f9abcf57e64 : SUCCESS in 46m 33s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a9b37f1aa5064cd49f35ead3f40afa61 : SUCCESS in 43m 35s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/b52e1c2da49041279e55edf37e581a82 : SUCCESS in 45m 40s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d0577bd0eea14726969ca04faac4e2fc : SUCCESS in 1h 03m 58s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/69799a3a05ac4dd6bc595a95e4e1bdca : SUCCESS in 31m 25s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/cc3d7b8880464861baeadf43ff231086 : SUCCESS in 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d1f26f4367b9504324d51583afe2d340b878d4ce","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-01 18:46:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"08c0cda05597d7730b6c3622c38529ec8bb18dd4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-01 21:05:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"166dae04e5fa09fd473776eb6d961bef9ac05579","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-02 00:25:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nFast approve this patch since it fixes the fullstack job","accounts_in_message":[],"_revision_number":2},{"id":"60c9d8986578686a36532c888e4b9f957460e71d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 00:25:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a8796d58fcd23ffa048d500302384971b0efd88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 01:35:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"21cf3df1a4cff99bfcbac06576ccd7af6ec7eaf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 01:35:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb45d1e6fd2c46329e53ceebdf8ee1b5 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41daec297fe34e319309b0648f9c9f93 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc3b575acefc47e08e630041daabf355 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5752129964084f23877113385d035a76 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68ec4274d13d4c0daa238e0e0f5f4ae2 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1315d1c5d2f4c37ab646a363262dfbb : SUCCESS in 3m 46s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/c605da574b364afa91257ac148c10cd5 : SUCCESS in 49m 38s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/1b24eeb34804429a82e70fdb20e5d9b6 : SUCCESS in 46m 09s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/86c26d5404554d0b8b7e512f1a163ba3 : SUCCESS in 48m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f66ce636218a426a80621a1751f0578a : SUCCESS in 1h 05m 02s","accounts_in_message":[],"_revision_number":2},{"id":"37b6447bc1106927444f1711f1f55d3b9d2b1e7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 01:37:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/721bafe6ad8f4c0ab01dc314f38657aa : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec96c5e96ba7ee329b88414e777dc8c01ad64c53","revisions":{"55c083b0e435ab7f4fc4e7731e6841d235d68861":{"kind":"REWORK","_number":1,"created":"2019-09-01 16:39:11.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/679573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/73/679573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/679573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/679573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/679573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/73/679573/1"}}},"commit":{"parents":[{"commit":"4479d5758cb584eddce05cf6615baeb03b9ace2e","subject":"Merge \"Add a job for testing zun standalone deployment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4479d5758cb584eddce05cf6615baeb03b9ace2e"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 16:24:56.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 16:38:34.000000000","tz":0},"subject":"Exponential backoff on retrying create network","message":"Exponential backoff on retrying create network\n\nCloses-Bug: #1842243\nChange-Id: I00bd54b6d71b3e72499ff9e906f2d0a1cdf4e284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/55c083b0e435ab7f4fc4e7731e6841d235d68861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/55c083b0e435ab7f4fc4e7731e6841d235d68861"}]},"branch":"refs/heads/master"},"ec96c5e96ba7ee329b88414e777dc8c01ad64c53":{"kind":"REWORK","_number":2,"created":"2019-09-01 17:21:48.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/679573/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/73/679573/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/679573/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/679573/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/679573/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/73/679573/2"}}},"commit":{"parents":[{"commit":"4479d5758cb584eddce05cf6615baeb03b9ace2e","subject":"Merge \"Add a job for testing zun standalone deployment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4479d5758cb584eddce05cf6615baeb03b9ace2e"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 16:24:56.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 17:21:43.000000000","tz":0},"subject":"Exponential backoff on retrying create network","message":"Exponential backoff on retrying create network\n\nCloses-Bug: #1842243\nChange-Id: I00bd54b6d71b3e72499ff9e906f2d0a1cdf4e284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ec96c5e96ba7ee329b88414e777dc8c01ad64c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ec96c5e96ba7ee329b88414e777dc8c01ad64c53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
