)]}'
{"id":"openstack%2Fzun~679590","triplet_id":"openstack%2Fzun~master~If025a9d3f8e2708c1bebcb2419dd840b15cab79b","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"If025a9d3f8e2708c1bebcb2419dd840b15cab79b","subject":"Fetch docker logs in post run","status":"MERGED","created":"2019-09-02 04:03:18.000000000","updated":"2019-09-24 02:00:44.000000000","submitted":"2019-09-24 01:58:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679590-1569290322429-f5fe0389","meta_rev_id":"84a0aee3b122ca427cbb01b9b2b476d7c35557b0","_number":679590,"virtual_id_number":679590,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-15 17:08:15.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 01:58:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-09-24 00:41:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-09-24 00:41:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 00:41:22.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-09-24 01:58:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d87e79c82e4273dad5c12f59a75dc5cf0c975519","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-02 04:03:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3832770da3749e426975835cd1c46112514f6d4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 05:01:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5d594d3baed4aadbf8535ac6c722c03 : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e018b17a026042a48b3ae8ec7a8183c7 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/499922e1854c409f8d1ddc5d631eb389 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64872759cd0f4a81996dec43962f8731 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f11352977e0454c80ce53cafe001f8b : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afdde8dc977242358ae61ac2a5992f92 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833fe4bca4fa408880a125c7c354f7fb : SUCCESS in 4m 38s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/8e1aa15109f544459865e6a9aa775f49 : POST_FAILURE in 43m 56s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/e0b926df37cb47f89a8b7f855405f207 : POST_FAILURE in 40m 44s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/8e821acc0f484a82b72e2d17b420f4c4 : POST_FAILURE in 40m 57s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/4ed9f05cfed14940a4c1f94132a38dd1 : POST_FAILURE in 55m 12s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/dc848564f2f94ea593512719fd933c1f : POST_FAILURE in 29m 08s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/973064fe819940939831e86f34326346 : POST_FAILURE in 40m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d5489778a6b93e5a208d529889b19bba7c06e3d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-02 15:15:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16e65a93ae7c518456bcee1cb000907258cc16ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 16:32:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac6436904f194bf39a74a46ebaf62176 : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d42f32e7f8de43c69249dea18ff14fad : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aca1f442fa1460caabd96ed6024a7d1 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b50070f27a484f5790eb74a0ff79632b : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c6c70fcdcae4711bf5137ba66ea55b5 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0900427cd4a428aa6fe62846fcf3589 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be82aa363cb240c7a8e8c66bd0da4c99 : SUCCESS in 5m 20s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/1408a606afa041f3bfdf4400a5ef1b98 : SUCCESS in 52m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/c6ac51716af04d869325c76eb7def67a : SUCCESS in 47m 34s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e2cbcbc740e64105be9d1a670a6e8586 : SUCCESS in 44m 46s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/0dad69545cc044c7a8a61c7960b212e7 : SUCCESS in 1h 04m 32s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/2cf5eccad5c34408b6e9bbfc8a6e8b43 : SUCCESS in 48m 04s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/c2f7f794fa7d48b8b4c787ce45a642b0 : SUCCESS in 42m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2636f11fbd38036f277923453e4c3b90ada97d63","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-13 02:28:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e8803c67291030ff38b0bc803203b9e35c78614c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 03:47:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0e987e558f048afa2edc909f926c7c9 : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca2dad7e79c744deb1bc2f8e0b6ff9b9 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/917066e7c60e406e88e8f5be39f71449 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56f6ffa93de54c24a69d16b0e04a5554 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58389b4003064332896159b87a9daebe : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/790f117adfa1462494242b8893b2a18f : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e97a08c7acc84218aaa3c8c7a2e8de91 : SUCCESS in 4m 07s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/0b354d36b2ed45feb7d3d5d7a923672d : SUCCESS in 58m 54s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d9276bf18fd74fe9834a5fcc0571740c : SUCCESS in 47m 16s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/1d836b9793fb4fa5b85e307b636778d9 : SUCCESS in 46m 53s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/1959b953bd274440ba2439abfeebc871 : FAILURE in 1h 03m 47s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/51617b4901764b8ca2a4153c069c34e2 : SUCCESS in 30m 46s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/f6f849c3ca7d4e82892e4886efc6c17c : SUCCESS in 43m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac1228eeb14046a39d7469f7f32c364e960d2b51","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-15 17:08:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"595c767c2c47a88da26564b21ae6fb81913822b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 18:09:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6421ee1374db41f391078b455b821f12 : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77a500397f0d49619bf60140f900378e : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82aaa5568e4a4566af8c311701829b6c : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a58184c4e43d4072b0fef5bb00b4be11 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f5d3aae6d8d4cd6b198dfc28ae8522b : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f1711b4f5dc40308202b7cb6dfde87a : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eec7c07192e4c8ca4b9df56b6182410 : SUCCESS in 3m 59s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/294a3dec322f46bc86f8fadd8d9cbe4b : SUCCESS in 51m 12s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/1053b7d5d2894140aad789d29e02c215 : SUCCESS in 43m 34s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/9f4d112be11a45e0b650b2822ea405a7 : SUCCESS in 1h 00m 22s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/538746cc69124e2c971fd8659dcd6090 : SUCCESS in 55m 44s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/7952fc15651545a4ae1c7dedbe9fb420 : SUCCESS in 27m 47s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/d7563c55a6f04793bcd699a725e6c896 : SUCCESS in 42m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"436c64f70fe82394fd4ab4f2cbe1b97b42df5186","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-09-24 00:41:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e465ae4c26a2dc6bf47bb7c2d8758a49f6e19350","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 00:41:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fa552664ad0c6150ee0d4449f73c78665f783a77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 01:58:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ba46c0f07bc440d8b9bc5126fd9822d : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2b63b7c9df447fd9fad446116274b39 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f8b41235379e45979a5414f657a72f98 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92f2947d6cd34cc0a5c82dfa49768a33 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/277678799b3c4f76a675b8a609aadff0 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79c26aed45144ec89a9e0bd3282e7da : SUCCESS in 4m 33s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/73dbbc2a0faa466bb5a9215a1f7cd011 : SUCCESS in 52m 59s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/837e8939075c4b86a270690bd8eeb9c9 : SUCCESS in 58m 30s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/de5d65be1225454e80cac920b7fb9352 : SUCCESS in 47m 02s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/871cbf2c40844582a14e18e93b5cb768 : SUCCESS in 1h 11m 39s","accounts_in_message":[],"_revision_number":2},{"id":"30f03354cb38479d3c020572956abf227101dbeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 01:58:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"84a0aee3b122ca427cbb01b9b2b476d7c35557b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 02:00:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da7f9f319fae4c33861f37925cb82f97 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"87a0764cfb2a541484c3b36b52716579502410e1","revisions":{"e00d8209c16d664df652baf3259282ddf2252863":{"kind":"REWORK","_number":1,"created":"2019-09-02 04:03:18.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/90/679590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/90/679590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/679590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/679590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/679590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/90/679590/1"}}},"commit":{"parents":[{"commit":"ec96c5e96ba7ee329b88414e777dc8c01ad64c53","subject":"Exponential backoff on retrying create network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ec96c5e96ba7ee329b88414e777dc8c01ad64c53"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-02 04:03:00.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-02 04:03:00.000000000","tz":0},"subject":"Fetch docker logs in post run","message":"Fetch docker logs in post run\n\nChange-Id: If025a9d3f8e2708c1bebcb2419dd840b15cab79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e00d8209c16d664df652baf3259282ddf2252863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e00d8209c16d664df652baf3259282ddf2252863"}]},"branch":"refs/heads/master"},"87a0764cfb2a541484c3b36b52716579502410e1":{"kind":"REWORK","_number":2,"created":"2019-09-02 15:15:37.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/90/679590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/90/679590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/679590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/679590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/679590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/90/679590/2"}}},"commit":{"parents":[{"commit":"ec96c5e96ba7ee329b88414e777dc8c01ad64c53","subject":"Exponential backoff on retrying create network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ec96c5e96ba7ee329b88414e777dc8c01ad64c53"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-02 04:03:00.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-02 15:15:32.000000000","tz":0},"subject":"Fetch docker logs in post run","message":"Fetch docker logs in post run\n\nChange-Id: If025a9d3f8e2708c1bebcb2419dd840b15cab79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/87a0764cfb2a541484c3b36b52716579502410e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/87a0764cfb2a541484c3b36b52716579502410e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
