)]}'
{"id":"openstack%2Fzun~685613","triplet_id":"openstack%2Fzun~master~Ie90e1ebde833938662e2ea3c5f953ed57db918a0","project":"openstack/zun","branch":"master","topic":"bug/1845650","hashtags":[],"change_id":"Ie90e1ebde833938662e2ea3c5f953ed57db918a0","subject":"Fix image search 500 error","status":"MERGED","created":"2019-09-29 15:54:43.000000000","updated":"2019-10-05 15:23:52.000000000","submitted":"2019-10-03 04:38:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685613-1570077511336-adebb769","meta_rev_id":"0b8f66ab2c34f5b0ccf44cffc980c7e763a3c20e","_number":685613,"virtual_id_number":685613,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 04:38:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-10-01 01:55:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-10-01 01:55:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 04:38:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"818d70b656d18f506f81328013662618aa8e67dc","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-29 15:54:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7345845782217ca0417f095daab5af026fc672fb","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-29 15:55:19.000000000","message":"Patch Set 1:\n\nBackport candidate","accounts_in_message":[],"_revision_number":1},{"id":"bf33cfbbaee62560de98a24d5663b2bd7781a2f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 17:02:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/845f2f942f934390b0ec8e48a6193fb9 : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1500fb2b7c940eba33859e256cd49ac : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/629261b404914d6bbd62564993f45504 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/396eb833b59a44e6bd25e25a9b0a000b : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa82b32f354c4529a5bb010bcea07881 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dea6ebe8848848feb46beff2697f749f : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c77504b4b2e4247a2fb016d85425adc : SUCCESS in 5m 22s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/1a9edc0545a34414b4d3314161a392ab : SUCCESS in 50m 40s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/28417d279cc9465887cd4e318cbaa4f5 : SUCCESS in 46m 08s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/bf4f282a758749fea8a6a8cc5d6c2237 : SUCCESS in 45m 15s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/620cc6ad3fc54bb29aeee74077d13411 : SUCCESS in 1h 01m 37s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/fd6cd25d716f49a2895cf451a6b8e9c6 : SUCCESS in 32m 07s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/fa8fb24fd901410bb9c904454390e352 : SUCCESS in 41m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fc3a62c4e136b65a7d70236089ad33b72f0fb13","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-01 01:55:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3156414a864dacddb3a5014ff499e17bf0b6282c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 01:55:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b4fb9d6bb697fdd993c4127feb4e5ab531e998d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 02:55:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa944a3f058a472e983819bb47fbefab : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62ee4bf33e646c0ba74e551244a681a : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc4e90e9039b46e29c93229e80bdf737 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b40aec3ddcd4acab7aaa829ce42380f : FAILURE in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f02b559511fe4e2f91374ba598c1c644 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f92aaa9dee5847108958cd5406fc5276 : SUCCESS in 4m 21s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/a3cbb811b29e4c648b2ee90ed1be8a87 : SUCCESS in 49m 53s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/63865f46817b4d3f90e7d5ab5a0ffeff : SUCCESS in 54m 36s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/56ff7c7abbaf473cbf15ec1faca555a8 : SUCCESS in 55m 55s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b4c77bf0c3584c8ebbe6bc40421ad5f9 : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d2d744dde41c628f8a0ffc9b1d68420f40b2b432","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-03 00:54:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a5297022b1e9473c27967db755da31ab94a2dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 02:09:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42f8106d3a2f4ab79ce1c3a6e4c47ae0 : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a067cbe768dd49f08782bda22b2c5b57 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09d10996c9941189cbc3ab55984f3aa : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fea92ec28e734ccc97170bf32d3ddd9b : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1369f524120c4f42914249ba3bce88b8 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79d5689af7a4445b899b0d24d4753919 : FAILURE in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e648f0ceb61465ca69aad9130d7dcc5 : SUCCESS in 3m 40s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/6e0e20d02acf48d38a5b74067f2a848d : SUCCESS in 53m 40s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/8a8d8bbf8d2c4a8ab5d71fbd23d13a09 : SUCCESS in 38m 51s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/09458af771044019b8f87db8c8817a41 : SUCCESS in 50m 35s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/bc1107608b124b47b4182c023751b19d : SUCCESS in 1h 09m 24s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/a1c573b12d504523a148d4b9110ddf9e : SUCCESS in 39m 12s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/52049854321c42e19145d4275ec2f495 : SUCCESS in 46m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a4b4a25a36cc0409aacd85c0c59b46e684af387","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-03 02:27:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d3792fb4d5d078009bca87cc22b6f482e052676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 03:42:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06941bc87cc24365ba1ed5d6dda97c14 : SUCCESS in 7m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57cb8161b9904d8c8917ed2045cc41c5 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84afa4aa418a4843a2301c1042867b6f : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/228e0adb0136473dad7c7891f0980412 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f76b6ccd11b46e193e5ce5d0692fa45 : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe34bc3657304a74937ceeea45dfb685 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc357f1d7a54b86a1827110f7a28844 : SUCCESS in 5m 01s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/018c68e68f3e416fadcf57e1280cbaed : SUCCESS in 1h 00m 49s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ceeedd3e77e446a8ac1d4100cfd1c309 : SUCCESS in 45m 41s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/c7961ec6fed64f41bec3e4e062be3f55 : SUCCESS in 56m 25s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/31fe0001324d441bb2b5c0727d14bc4f : SUCCESS in 1h 05m 16s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/6626ae5b269946fca879c18cf91fc0ca : SUCCESS in 45m 21s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/6e8257774e6542f5a8cca94616b41a6b : SUCCESS in 50m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5aeaadb44f3bf6392466cfc829e80db525a4aec0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 03:42:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12e945c921b81e9625488510005f20264af8a71d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 04:38:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a423d4c35e964128952f9e71115186d8 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/270bdb1cac924a04bb5d1daa49c1ba40 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d65cbe968c5342aab758f43f426643d7 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e79696509b8b433cbe5b8814f15c3b2c : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ffbf5e17658431798aad1116d7597cf : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daae5ed3534d41e49c90329f4c5e4a85 : SUCCESS in 4m 30s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/3c425319875744c98e23f211df19d5ac : SUCCESS in 52m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7846c300a3744407b772d1075519b525 : SUCCESS in 39m 10s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/0005c121f71240cc97598ae743e38db4 : SUCCESS in 43m 33s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/557b1bcad513473da5a57e64655c75a7 : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3ca6da94a205e3011c5a26e423230fd0dbae589c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 04:38:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"249e05557647885fb0fb2a08ef265e3caa0bdf38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 04:39:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78113dbfbd7f4f3e833499cbee18b585 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b8c79fecaa08394b2d88190b0977b7138d0abf24","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-05 15:23:52.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit cb0cb901bc76175b0ec7f0cda37e3883c2d6842d","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1712f184fcc621c9e32c845dba8d6dcf05f7f6cf","revisions":{"1712f184fcc621c9e32c845dba8d6dcf05f7f6cf":{"kind":"REWORK","_number":1,"created":"2019-09-29 15:54:43.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/13/685613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/13/685613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/13/685613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/13/685613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/13/685613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/13/685613/1"}}},"commit":{"parents":[{"commit":"ed7eabc94b481fc7afe1c87f5bb6ed806974cf92","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ed7eabc94b481fc7afe1c87f5bb6ed806974cf92"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-29 15:54:11.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-29 15:54:11.000000000","tz":0},"subject":"Fix image search 500 error","message":"Fix image search 500 error\n\nChange-Id: Ie90e1ebde833938662e2ea3c5f953ed57db918a0\nCloses-Bug: #1845650\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1712f184fcc621c9e32c845dba8d6dcf05f7f6cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1712f184fcc621c9e32c845dba8d6dcf05f7f6cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
