)]}'
{"id":"openstack%2Fzun~686898","triplet_id":"openstack%2Fzun~stable%2Ftrain~Ie90e1ebde833938662e2ea3c5f953ed57db918a0","project":"openstack/zun","branch":"stable/train","topic":"bug/1845650-stable/train","hashtags":[],"change_id":"Ie90e1ebde833938662e2ea3c5f953ed57db918a0","subject":"Fix image search 500 error","status":"MERGED","created":"2019-10-05 15:23:52.000000000","updated":"2019-10-25 15:34:10.000000000","submitted":"2019-10-25 15:33:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686898-1572017582542-edb63827","meta_rev_id":"c071f6fc20a90ae4c6f56a915f89b64f287b965e","_number":686898,"virtual_id_number":686898,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 15:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-10-05 20:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-10-25 12:40:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-10-05 20:28:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 12:40:35.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-25 15:33:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f396a8d1da5d99ea36a0ee1a8e4ea584f7bbdac6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-05 15:23:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e74d07a56363e85301e703f8912a2286b6d2b8c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 15:38:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f7456b8d8ed477d83c370fcbc0982a7 : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed2841e2bb314f588c3d9e56da1c6dee : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c93050ad6254d31b9eac29f7c914251 : SUCCESS in 9m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30b3a2a26f0f460c99a731cb741dd1ea : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0feb6c3200624fda96b81cd674336a99 : FAILURE in 7m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a1f32deef024677a3a750b9e588aebf : FAILURE in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/345443a444474287b0c9e8e111aa87a6 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7606d94c95a58a3c69e297ef028137e429ebd344","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-05 19:34:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d8d596e973fbac5393ddc856b33da6e50bd25f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 19:41:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70ffc4f7758947eab22697d08e26b9ba : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e099847cec9b4d00ba8a7dc906d212c6 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f917d62bd040421ea591ffee90bc8b0d : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8982748539b74fbabb0ecec8e049fe4f : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50b78f7acb4a49fc91b2a691fe9b9339 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/088abc0e48804e7b9ae579394366e332 : FAILURE in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a496f74eda347efb712db09eb18243d : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0e6a7be3c7793014a2cb051ee771c2b56bdd231d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-05 20:06:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"518cfd25426c8074b9d6c627f8d6826fa3d19547","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 20:20:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56937918f840416bb8bd9a4f3b9ca7fa : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f6b086047ab42078f2f44b820af4d82 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea91168f7b24ac19877c51872209cc6 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38d85d82b8934249891f6225ccaa99e6 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbd5ae315f1c480d93a50e6b90140149 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cefb303ee9ec4c7498a37eb5c73d756f : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b36dbb46ecdd4c598be8954170b6213b : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7905d8583b1d3208d30f3eb7db51d3f042471ad4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-05 20:28:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"af003fdb63888c0a1816ec5f04e2d99fcdaa13f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 20:29:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee70a0bc912af5097857707453fb545c5b71c864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 20:38:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96cf4554d7154ae19c91bc11f3c4d697 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72829ad4a92342b387011778c4e87e0b : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad08354c5f6c4915b75560afb53e85a3 : FAILURE in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49cdf95a86e24ca5a6ec0aba08b0237c : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f5c4f5225e14d0ebfebd375d6af4244 : FAILURE in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae092e9688a4b34a66cfdb5d1a414ed : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"56b50950641c226d1279f9d5b21df63da54d55eb","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-05 22:05:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2527789be880139c2c1331d54f3b48759bdff300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 22:14:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38ee333d68024536a9535855a8e1ad41 : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b937745f9e944b1fa525014df6704423 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e410a90a9045c0b24b47b1643df0fc : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f877e644bf70447298be05b2bdbe68a3 : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cac77f620ca4bd4b5ce39eb20ba71da : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e73d51904e4a45d0b72e161328d9aa19 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/266a6f1d46194ee1912dd5a297578cc3 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7cd789ebbd64b55dd2c657289bee9ecab3951730","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-06 01:33:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ba2090aa7c79ca8fe1a9ed2190289a880f87916a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-06 01:39:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce21e42dd3374a198ffafa54ff06cdd4 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f179dac9caf14088b1c33e620660abcb : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd4d79891674b26b43618633653490d : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64365dd08796435ba350e14711aeb9ff : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ec32bced0d14e42bcf378058b50e019 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78734abe00ff4e6ab0d00990dc1dcf40 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c54d2bcbe50423f91bd85f15d87bcff : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"fc5fa293428df10f8f74ab85f9383e06acf359e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-06 01:39:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc1778fde081f7cd0a929b8331ac21f30128ecc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-06 01:47:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7badd07e4764d2997df72ee2a487507 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2001af09ce054baf8856a2c7ff5b0fd2 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da80731df2e34c2db2158da56be657ea : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04ac9be1f22e45b796d3f71230e30295 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/754ba2c98afb42e2a3b14ab6507f602d : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbede6c59c6448d998ebecc3370e8a4e : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4fda894da9311685e2bada9760e10d551bbab03b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-06 03:29:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b294bdba55e7d1f932526203b274867bb22a81aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-06 03:38:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/616c66d1280a4afe80474e4bbc49ba53 : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33ed49a906d94311a93aa2343512070a : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70510e4808314a9a8c794369ce6e3854 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47d2c8f379f0454ca2f16ff16f6a9656 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/044cb6f9688a43a9ab46b267de2b535c : FAILURE in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/624017b198de4b3e9bb6bdd0572e5dbb : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e34077d6d4304d5aa210ae679b8f6287 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d36782098ffecf34841408530abb0b7d918c8758","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-07 02:52:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75f94cf3ec7d4e5d8d9b7f3273d5cf8d420fde0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 03:02:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc2a9ab2b2b64f6b8bfd9ae2dc616c98 : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/352d1e8280584a2f999f9a5703fc09d2 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bef4f92a1a9d478fb41212dcba12d50d : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2588801405f34668a2127db9463d4986 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4a06924c2bf472882dd138b1bb05bac : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d1d90cb37fa481ea76404e265d2e0ec : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66098735e7ae4e8db4ec231c8e05725b : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"404389f5efc21dd9690194a829ec7fb2b93e1100","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 03:02:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b442671eeccc24f4ed3d103c16c5d0cf80c27421","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 03:09:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c5ca847ddd944ed875f5493afc15957 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc89f802541f42ce909ff4a16040b7a8 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9c2be483952488798076a5bb9e6bcbf : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ac46b5b4a38444db162e7849a9c2412 : FAILURE in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b2bbb83561d416ca94833fc080d9fea : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6539b8bfec461f96c5cbd69acd8282 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b7d113716f74519fbc26fb82575b65dc9087ae46","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-12 23:08:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"78766d50d76950da97dda9c6bcb4839fd9051a10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 00:08:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80d2ef7fa2874632a8ad995216c292cd : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48a0ccaa638645e0b40de49259e18746 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7507914aa16749b79da60b3f314a68da : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b729db89c1c24e2a8229ae7038d9d512 : FAILURE in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdfa51c157734690b5b7c0c2c5dc5808 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe7a6cc359fe4c1fba5c761431283745 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1267753863415f8bb7409f2beb5d08 : SUCCESS in 3m 32s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/d2b97353a7fc49938bc0d2d35411d7fb : SUCCESS in 48m 42s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/71bad18db10949d5b4a981922d7f21cf : SUCCESS in 45m 14s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/1bb3f739e823482c929ee066de887036 : SUCCESS in 43m 00s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/c5bb351c45d945b88eb464025b011753 : SUCCESS in 57m 45s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/f2a46f0d7eb04d24917b1ef336a7957f : SUCCESS in 29m 58s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/58feb06e10c448e49f6cfce01e94cd1a : SUCCESS in 35m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63a7cfd772719f153e612713d2d815add112766c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-13 01:31:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"53349ace13c19c593f7a1c567d96551cc00f812f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 02:32:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b7a2bf3ff3449719703824f0a7a8e74 : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ed44e3aaa914832b54aa00e94ff57dc : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30c7693c4134439e8cea9f535bfde1bf : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8f69e7233e64e5c9230c6cd4af48848 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/011d343d37784b5289986441f169e7b4 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c7291a567004be4aa6697a106add437 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/976a56290764484496e48dde7a5defe6 : SUCCESS in 3m 44s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/990c552d66b14068b61892df87893f6f : SUCCESS in 43m 47s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/87422e52834c4595960a7955fda9bcdc : SUCCESS in 45m 16s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/4a6ed884688a4adaa0dea2a592acb0b8 : SUCCESS in 47m 08s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/03c8adaf1dc04715aec80b42632ea758 : SUCCESS in 58m 27s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/e0b6c96c83f249dcac28b8858e438ee8 : SUCCESS in 33m 57s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/27406872f974421e8545e21738bcab2e : SUCCESS in 39m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abfe54e7b7d5a6538b403f2b31a592cbc32e2da9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 02:32:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05226ad4c523365e3114e22a4559fcffb966e35b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 03:36:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acb4f262b49948b1a7f41cae8c0317a8 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f5696297504a13b570a0c791904cee : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7259d7b1bcd4465593525a414bd352cc : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85b2815f853849e2a6b6e91c8bdf9f5f : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d728082614344d76917c5cb8b22fbbed : FAILURE in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6813bb83269942909e1c9704620b8150 : SUCCESS in 3m 28s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/04cb7f263481411dbcb06965ea408210 : FAILURE in 1h 02m 04s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/80dc72725f8d42fea11e8aebc8566312 : SUCCESS in 44m 14s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/0481a75621d54515873acfa9a222e5d2 : SUCCESS in 38m 42s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/6ca8ee72767846178b3bd1b3bee946a8 : SUCCESS in 50m 39s","accounts_in_message":[],"_revision_number":1},{"id":"383fc11178ff8dda81852b99161975d2c9e5fb68","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-25 12:40:35.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8d6dad80fdfb9a72244862581d6cde492a06ac95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:52:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46aecf551fb342ff8bcd53b669eba62b : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e3a4528f3ea4cc4a8332acf27cc8834 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f22eeee4bb2546ea9c0caa1e5763f794 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae3f20a222c64b2db888adf2068c2e06 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43355f0d4bb44763ab21674034ac8f27 : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c9a497e819d43b780209c6dd9d0b4e7 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a28cc2d831914d4aa1eded0182553412 : SUCCESS in 6m 19s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/333dfc632a6b4556b1039f5645217aef : SUCCESS in 53m 36s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2c3bc5abe430439fb17fa30b6a9574f2 : SUCCESS in 52m 05s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/a59e3e39984b449fb021dc2e20275dea : SUCCESS in 51m 42s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/bb9f8030ed15466586f25e13e208e33e : SUCCESS in 1h 08m 57s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/ad38903384d0432da14e44f4684365fe : SUCCESS in 34m 35s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/e9ee41d7a69240bfb35f08d2e532a9cd : SUCCESS in 42m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70bc949ae53186da5f605f2ee54e86d0bf4685da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:52:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"239e9fbbc05ca0ada37fd3ba8b71baca02462da6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 15:33:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/112cf6382eb84530bdf198062e922533 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8834334e40b41a39fd20e14bfa25d6e : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9bc6e86b05f54babaed230bae3912645 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f04130d80614577bc54cdefe09d2b42 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/216a1bbcdfeb4335a504ddadb1f09bfd : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c5418110ca4b0a9257e5672d812566 : SUCCESS in 5m 40s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/6451fa1c644548318a9e7c564c7515d0 : SUCCESS in 54m 35s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/dbc8c49026c04b249faf777943859775 : SUCCESS in 44m 21s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e916793ff8c944e3b8fd96e0e8e3b9e4 : SUCCESS in 46m 12s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/fa88e6f3f6bc46869f099ff1b32e37c8 : SUCCESS in 1h 06m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0ada7c1595100fc8fa4f8200ded349c481f1c09d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 15:33:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1c5505a1fe1cb9a79256c0e434e5cd07739b2ea3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 15:34:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca54ca79ab7048ae9f24f66aec5a218c : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb0cb901bc76175b0ec7f0cda37e3883c2d6842d","revisions":{"cb0cb901bc76175b0ec7f0cda37e3883c2d6842d":{"kind":"REWORK","_number":1,"created":"2019-10-05 15:23:52.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/98/686898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/98/686898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/98/686898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/98/686898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/98/686898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/98/686898/1"}}},"commit":{"parents":[{"commit":"d0733c99317c9613c7fc3c98946ae9207f709d01","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d0733c99317c9613c7fc3c98946ae9207f709d01"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-29 15:54:11.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-05 15:23:52.000000000","tz":0},"subject":"Fix image search 500 error","message":"Fix image search 500 error\n\nChange-Id: Ie90e1ebde833938662e2ea3c5f953ed57db918a0\nCloses-Bug: #1845650\n(cherry picked from commit 1712f184fcc621c9e32c845dba8d6dcf05f7f6cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/cb0cb901bc76175b0ec7f0cda37e3883c2d6842d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/cb0cb901bc76175b0ec7f0cda37e3883c2d6842d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
