)]}'
{"id":"openstack%2Fzun~688285","triplet_id":"openstack%2Fzun~master~I3c5971635ba3a2fdf16f004524c824d21c19716f","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I3c5971635ba3a2fdf16f004524c824d21c19716f","subject":"Log more information on websocketproxy","status":"NEW","created":"2019-10-12 23:08:56.000000000","updated":"2019-10-13 04:21:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a61a2a569bd9952cf239cd8d18cf6ec7378efea6","_number":688285,"virtual_id_number":688285,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-13 04:21:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-13 04:21:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43b59792a7cd99f7df108b852d5770663d843b05","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-12 23:08:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53700ccb8ec100e849fa64685c977a052c944f9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 00:01:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1833d438d764e86bdc08c702d31bddf : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/095887e504fa440680ebd6cc23856238 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbab4ffb9d6548908a31a33ed02dee6b : FAILURE in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c7f859f9abd4790be84d38b543107b4 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adfb68c0d8754e96bc4c2f67d80b3279 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4005c14098774bd5b661c67114fc05f2 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb46f18f5f6d4dedb061515a04e3662b : SUCCESS in 3m 28s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/cefdb8099017469eb69960cdcec8dcf8 : SUCCESS in 45m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/e1fe6115cfca4e2e874d14ba10d6ff29 : SUCCESS in 42m 39s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/bdb48936958c4f8b9cacc0eb30d402c6 : SUCCESS in 42m 50s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2ae0712d51064a1f8de40279cfbd6248 : SUCCESS in 51m 37s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/34498401e69c4e93951126c5ac94a71d : SUCCESS in 29m 31s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/89af5c0aa06c4bc9a23ec7e12cd99538 : SUCCESS in 38m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81c84955bf8f29b0cb2f54eac7819999dff0f28a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-13 03:19:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a61a2a569bd9952cf239cd8d18cf6ec7378efea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 04:21:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/816c4270599c458ca036de660874b595 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d178549bd824bb683d453668753ce74 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7617e13d4b124a5393dc286216e08ba7 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52b43d5544764ef6a1412723b667ba51 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bb777255d77480c9d46dd1f59c29e77 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc7aa7b7210747faa3f610541c340366 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a960e2d638be4a60a04f13cfc3093baa : SUCCESS in 3m 58s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/ac710d6ddd854fd9ab44951e7107e439 : SUCCESS in 40m 32s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/325da7737c4e4d96b7f06fcdf53a03b6 : SUCCESS in 46m 33s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/0f7f288e1a414b299e74eb683450963a : SUCCESS in 41m 48s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d775a7cf1a2d4a35b834e741fd0f5fab : SUCCESS in 56m 39s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/389af5839fbd4f0caaa17f8a9464cba4 : SUCCESS in 25m 40s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/1be0277b6ed7453b9c5a56c5b15d820a : SUCCESS in 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6c4e1c1c6a7509b9b8f07c29c9cead46e9b81435","revisions":{"a8720a10e8975ec3e0479e19f23e4c57939f40de":{"kind":"REWORK","_number":1,"created":"2019-10-12 23:08:56.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/85/688285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/85/688285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/688285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/688285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/688285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/85/688285/1"}}},"commit":{"parents":[{"commit":"1712f184fcc621c9e32c845dba8d6dcf05f7f6cf","subject":"Fix image search 500 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1712f184fcc621c9e32c845dba8d6dcf05f7f6cf"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-12 23:07:00.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-12 23:08:55.000000000","tz":0},"subject":"[WIP] Log more information on websocketproxy","message":"[WIP] Log more information on websocketproxy\n\nChange-Id: I3c5971635ba3a2fdf16f004524c824d21c19716f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a8720a10e8975ec3e0479e19f23e4c57939f40de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a8720a10e8975ec3e0479e19f23e4c57939f40de"}]},"branch":"refs/heads/master"},"6c4e1c1c6a7509b9b8f07c29c9cead46e9b81435":{"kind":"REWORK","_number":2,"created":"2019-10-13 03:19:54.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/85/688285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/85/688285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/688285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/688285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/688285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/85/688285/2"}}},"commit":{"parents":[{"commit":"1712f184fcc621c9e32c845dba8d6dcf05f7f6cf","subject":"Fix image search 500 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1712f184fcc621c9e32c845dba8d6dcf05f7f6cf"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-12 23:07:00.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-13 03:19:42.000000000","tz":0},"subject":"Log more information on websocketproxy","message":"Log more information on websocketproxy\n\nChange-Id: I3c5971635ba3a2fdf16f004524c824d21c19716f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6c4e1c1c6a7509b9b8f07c29c9cead46e9b81435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6c4e1c1c6a7509b9b8f07c29c9cead46e9b81435"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
