)]}'
{"id":"openstack%2Fzun~691137","triplet_id":"openstack%2Fzun~master~Iec11d37d588943e6cd799b02f8a50c50c9de4387","project":"openstack/zun","branch":"master","topic":"bug/1849754","hashtags":[],"change_id":"Iec11d37d588943e6cd799b02f8a50c50c9de4387","subject":"Fix response of json schema validation error","status":"MERGED","created":"2019-10-25 01:19:40.000000000","updated":"2019-10-26 18:59:23.000000000","submitted":"2019-10-25 17:56:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691137-1572026171679-256045c0","meta_rev_id":"403aecb2e08d59b30ceedc05c7c1810416991404","_number":691137,"virtual_id_number":691137,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 17:56:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-10-25 16:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-10-25 16:49:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 17:56:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c702447460dd664c169aea664489e5eee20c68ab","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-25 01:19:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3f591f4fc831687730680ca54777edfe29e2104","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-25 01:20:51.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"529b711cf2f36268544267e8582efb891c1e25fb","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-25 01:37:16.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"475f974d4632ffea8fe06d223b087928ed46274d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 03:12:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/377530b331a54571bcd896c890428727 : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97996c49556f44eebdefbc648951ea66 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/918ab7edd8ce480f8a39f1192d3f53eb : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d120e8825d3a400fa0da934bdc824256 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c35fda56ca8f47c2810faf6d819306b4 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1baca80397e44081b52f4a7f12e222dd : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59a474616804ecfbdef4ff63575529a : SUCCESS in 4m 20s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/12e1fc6016ad410e9d4564e261c35ba0 : SUCCESS in 47m 17s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/8321c562c7e24274852851f6ece39521 : SUCCESS in 55m 35s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/b77313e0de8d4ada94372e349eb5a4ba : SUCCESS in 42m 12s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/114110828a06496a86c03e8a4a42d126 : SUCCESS in 1h 28m 03s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/f7a5114f6cd2449f843f1d3fa89bb352 : SUCCESS in 33m 12s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/d9413294175e49c7892c837f1294e2d8 : SUCCESS in 35m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ff4e102bb2b35c67d52f29bf3a085df7efae47b1","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-25 16:49:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bd1aa5c54955f96354b2fb39d4a8f18d1beac668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:50:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"84c0e066ede3fe4a6f809b3d8d020f574628f54d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:56:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3026b9856a1b9ad06e84c00a86e0e8f5992a0215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:56:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b66332e600545b7b1d0ea166f1df24a : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/284961abd9dc444f9fe9fd29f7484d0d : SUCCESS in 6m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eabdf641885a46cebd47190b32b853b2 : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d149dec8b6a3444b91931da90c6848f2 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c43ebffa82784e89870af2794487e04a : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e30ae2c3d3fd467a878d835a986559f3 : SUCCESS in 4m 20s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/63fe345a65d3475c93aef06975f06a8d : SUCCESS in 44m 05s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a90d8093b2e8459faaa44a109e770283 : SUCCESS in 50m 38s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/a3926ae9f4e74205a4d07235b16893eb : SUCCESS in 51m 00s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/75220c4850984e91987a3c72fd8942e5 : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":3},{"id":"f43a7c68c7ceb06ba3c75b2e9c8ba5ccdf300e38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:57:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c32135d3b6424d778fb82a8e3ac7009d : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3},{"id":"87aa091196876800328b9720db566314a4cb9570","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-26 18:59:23.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 924874ad87ac02f25b88da5e47e287fce3ffcba0","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"97f621c5a8f1a3bbce971702811bdb8a6bf1fa8c","revisions":{"eb450a85312e98ce95d3f0175da4d7ec0475c083":{"kind":"REWORK","_number":1,"created":"2019-10-25 01:19:40.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/37/691137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/37/691137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/691137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/691137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/691137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/37/691137/1"}}},"commit":{"parents":[{"commit":"1712f184fcc621c9e32c845dba8d6dcf05f7f6cf","subject":"Fix image search 500 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1712f184fcc621c9e32c845dba8d6dcf05f7f6cf"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-25 01:15:33.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-25 01:15:33.000000000","tz":0},"subject":"Fix response of json schema validation error","message":"Fix response of json schema validation error\n\nIn before, we returned a long error message as http response.\nThe long error message is not user-friendly, and it also causes\nthe CI unit test job to fail.\n\nChange-Id: Iec11d37d588943e6cd799b02f8a50c50c9de4387\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/eb450a85312e98ce95d3f0175da4d7ec0475c083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/eb450a85312e98ce95d3f0175da4d7ec0475c083"}]},"branch":"refs/heads/master"},"7413c9c1b7f949b459ce90e7f423efd9c6f1034d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-25 01:20:51.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/37/691137/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/37/691137/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/691137/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/691137/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/691137/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/37/691137/2"}}},"commit":{"parents":[{"commit":"1712f184fcc621c9e32c845dba8d6dcf05f7f6cf","subject":"Fix image search 500 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1712f184fcc621c9e32c845dba8d6dcf05f7f6cf"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-25 01:15:33.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-25 01:19:52.000000000","tz":0},"subject":"Fix response of json schema validation error","message":"Fix response of json schema validation error\n\nIn before, we returned a long error message as http response.\nThe long error message is not user-friendly, and it also causes\nthe CI unit test job to fail. This commit returned a shorter\nerror message (ex.message) instead.\n\nChange-Id: Iec11d37d588943e6cd799b02f8a50c50c9de4387\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7413c9c1b7f949b459ce90e7f423efd9c6f1034d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7413c9c1b7f949b459ce90e7f423efd9c6f1034d"}]},"branch":"refs/heads/master"},"97f621c5a8f1a3bbce971702811bdb8a6bf1fa8c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-25 01:37:16.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/37/691137/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/37/691137/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/691137/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/691137/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/691137/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/37/691137/3"}}},"commit":{"parents":[{"commit":"1712f184fcc621c9e32c845dba8d6dcf05f7f6cf","subject":"Fix image search 500 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1712f184fcc621c9e32c845dba8d6dcf05f7f6cf"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-25 01:15:33.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-25 01:36:52.000000000","tz":0},"subject":"Fix response of json schema validation error","message":"Fix response of json schema validation error\n\nIn before, we returned a long error message as http response.\nThe long error message is not user-friendly, and it also causes\nthe CI unit test job to fail. This commit returned a shorter\nerror message (ex.message) instead.\n\nCloses-Bug: #1849754\nCloses-Bug: #1849752\nChange-Id: Iec11d37d588943e6cd799b02f8a50c50c9de4387\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/97f621c5a8f1a3bbce971702811bdb8a6bf1fa8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/97f621c5a8f1a3bbce971702811bdb8a6bf1fa8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
