)]}'
{"id":"openstack%2Fzun~692483","triplet_id":"openstack%2Fzun~master~I48463714ce0e1519c1f8c1def15d38b1395238fe","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I48463714ce0e1519c1f8c1def15d38b1395238fe","subject":"tox: Keeping going with docs","status":"MERGED","created":"2019-11-01 03:39:26.000000000","updated":"2020-05-19 03:40:09.000000000","submitted":"2020-05-19 03:38:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692483-1589859536277-f8705a57","meta_rev_id":"c14b851a91e7008f54bf2622afacc7bf70d6cb37","_number":692483,"virtual_id_number":692483,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"date":"2020-05-18 01:40:05.000000000","_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-19 03:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-05-19 02:29:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-05-19 02:29:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 02:29:29.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-05-19 03:38:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7700d2b63cce90182186a8c257e151e7e95a2d5","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-11-01 03:39:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c154072e635e96e84dc4548180a8b032f0626755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 05:10:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/41b4a51a3cae4d468603105bf05191cf : SUCCESS in 6m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/37142bd18bc3452b882ef0368ce28e44 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c29a1037c69d4e0baccf2585c66c22dc : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c06ec27ce66f4a008218d434fae8fafb : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bc41c7556634ea199c00a28c45bddec : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/806067ca764c4c948826db6ed360ffa1 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ba6892bd5104fc7b46de588b06525f7 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cc96eb96fb84987bd2ae1e839685900 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f69a3811c6a425b9662d40dcf2df900 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2b5ba6409a9482286ef15941a087092 : SUCCESS in 5m 30s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/dc0a719a3cfc4c94a11e01520cb140ae : SUCCESS in 55m 10s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/55b6d7eb7af848c2a7ac8f9f6e5ac694 : SUCCESS in 42m 16s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/28a4e058014548eb95c8dfe0afb0955c : SUCCESS in 41m 33s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ef8b45b53742446a819161c5403ad8fd : SUCCESS in 56m 48s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/b22edbfb02934754b0041125c03eb4c4 : SUCCESS in 36m 00s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/b2955dcff471422dad4ecda0564dc57c : SUCCESS in 38m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40badd316cb470bfaef9662033199e6f22fed1b9","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-17 14:59:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"665a70e72ad7302926d8dc2b4ca84bc545186ae6","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-17 15:01:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"288b827ba9dd249040506444d1aa9827f5c67546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 17:03:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7945da3f235e475fa1f1ddcb3e058c47 : SUCCESS in 4m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4fdeacced52d4d1293b99549fce1399e : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02b057330a2f435c805e7d660bbf422d : SUCCESS in 7m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d263f7d29c14d0798607a3482812355 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ecdf1338624e369148231e588ea3b8 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90a4f704c8734c3eade855040eb4c4cf : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/041ee885db594caa8d94f16560274b9a : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9afaa3a1959947ed9f83d4e4046a22f7 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd30439df7a943ffaa6e4854f29c6542 : SUCCESS in 4m 44s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/e78c58d912d143bd910a8346bb75b3f1 : SUCCESS in 50m 20s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/c1e4246e6827483bad97c5ac88fc28ad : SUCCESS in 45m 18s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/251d95d9bfd44e6c966b56aecdacb5e6 : FAILURE in 1h 58m 19s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/640e04e1bdbd4a19a25ee5c0915b7432 : SUCCESS in 33m 28s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/0a95c58542e14e118987b3f23f8f39d2 : SUCCESS in 39m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"665a9f603b7ba4ef111aba4812d1f59a1523e3e1","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-18 01:40:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d63cf8b02e4ddab2e2cd55e7b5326506d6ae24c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 02:45:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/47c3fc4669344c41bd1441057f42cff9 : SUCCESS in 4m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d8566adbdf4c452aaa290d63288137e3 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de93c9a25a234c5da873e0cc3ecb6aba : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4792ce9b08504c02a5ca43ccbf496e20 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36b9472aa70e4172a2937d1f643da597 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0bf1a7c97934c3a8b8e07633f440a2d : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eccb11a9343746d1b9b678d34a047f23 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a4cf1688734eb1b3d310f4926910cb : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13051dc467ef49ce8053ac0c2f9ea8e7 : SUCCESS in 3m 09s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2be8e7dfb09b4bb78c1f4b52e29bd98e : SUCCESS in 47m 41s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/1b401551f793445c89f51367b5eef73c : SUCCESS in 47m 33s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/829b2d6eae9a4e5bb70cf1fdea3bef25 : SUCCESS in 59m 34s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/5968dab0f25547339f29747fb8c79581 : SUCCESS in 34m 12s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/1be0a79d7faf43a48c6ef8723d6400c7 : SUCCESS in 41m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f719a94d914541a6b56689b3fba7745e1b0eac6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-19 02:29:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1e886c67ee8c79088bd94d1b4171caa8f4c7452c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 02:29:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a335ddd3192761c42083e0d754a6ca61fd4dda6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 03:38:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8580cfe4bb634e35b5db4b8a1a099edb : SUCCESS in 4m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/112b8a22e05340078e92c3ba4667b6ef : SUCCESS in 2m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d1d0d96bf5d43f0a8ecd500b9e3c9ec : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a75857bb81854322aa5c23e55dbaf9e2 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef11096f310b4a00a614b6188596f287 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e556d20dfae64a17bc073d8984459e0c : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5abcedd683e414bae3a90682ab72e36 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf741c29f948491b8dcd696c0d2fe41d : SUCCESS in 3m 55s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2780667da95f4257bd6dea4abd95e8eb : SUCCESS in 49m 53s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/d7fab782d8c34027b2ed33e5ccf4f773 : SUCCESS in 48m 59s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/716cdd031a9c439a95203c4f9f115816 : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":3},{"id":"a7ce474b64d3280c60058de4e7fa6fa56931e5d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 03:38:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c14b851a91e7008f54bf2622afacc7bf70d6cb37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 03:40:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/8b37ecf7ef2a453da8988f40bef11b97 : SUCCESS in 57s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d040965379524fb19213c2474900266e : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa96a9cb93a14ff69f9d135b3cafbb65 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"42b7fb8b95edd0db601d92d0abd6c56f6f33902f","revisions":{"33796652a45602328fd587240443ba2262c6cb73":{"kind":"REWORK","_number":1,"created":"2019-11-01 03:39:26.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/83/692483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/83/692483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/83/692483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/83/692483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/83/692483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/83/692483/1"}}},"commit":{"parents":[{"commit":"1712f184fcc621c9e32c845dba8d6dcf05f7f6cf","subject":"Fix image search 500 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1712f184fcc621c9e32c845dba8d6dcf05f7f6cf"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-11-01 03:39:08.000000000","tz":480},"committer":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-11-01 03:39:08.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nChange-Id: I48463714ce0e1519c1f8c1def15d38b1395238fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/33796652a45602328fd587240443ba2262c6cb73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/33796652a45602328fd587240443ba2262c6cb73"}]},"branch":"refs/heads/master"},"9838ce34d592d2042235ec3d0accd2547d910a64":{"kind":"REWORK","_number":2,"created":"2020-05-17 14:59:50.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/83/692483/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/83/692483/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/83/692483/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/83/692483/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/83/692483/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/83/692483/2"}}},"commit":{"parents":[{"commit":"e96e4e81bc26397dc4b33016f769e8ec61e7c254","subject":"Merge \"Remove the unused coding style modules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e96e4e81bc26397dc4b33016f769e8ec61e7c254"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-11-01 03:39:08.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-17 14:59:20.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nChange-Id: I48463714ce0e1519c1f8c1def15d38b1395238fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9838ce34d592d2042235ec3d0accd2547d910a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9838ce34d592d2042235ec3d0accd2547d910a64"}]},"branch":"refs/heads/master"},"42b7fb8b95edd0db601d92d0abd6c56f6f33902f":{"kind":"REWORK","_number":3,"created":"2020-05-17 15:01:22.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/83/692483/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/83/692483/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/83/692483/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/83/692483/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/83/692483/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/83/692483/3"}}},"commit":{"parents":[{"commit":"e96e4e81bc26397dc4b33016f769e8ec61e7c254","subject":"Merge \"Remove the unused coding style modules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e96e4e81bc26397dc4b33016f769e8ec61e7c254"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-11-01 03:39:08.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-17 15:00:52.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nChange-Id: I48463714ce0e1519c1f8c1def15d38b1395238fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/42b7fb8b95edd0db601d92d0abd6c56f6f33902f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/42b7fb8b95edd0db601d92d0abd6c56f6f33902f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
