)]}'
{"id":"openstack%2Fzun~696780","triplet_id":"openstack%2Fzun~master~Ie70ed401dea63e9397d98c82b05059212e4e31c2","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ie70ed401dea63e9397d98c82b05059212e4e31c2","subject":"Change default of resume_container_state as True","status":"MERGED","created":"2019-11-30 22:18:58.000000000","updated":"2020-01-01 18:06:28.000000000","submitted":"2020-01-01 18:05:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696780-1577901901883-a23e576b","meta_rev_id":"d25686d08dcdab70bf2caa4693ee97dfc6f33332","_number":696780,"virtual_id_number":696780,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-01 18:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-12-30 17:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-12-30 17:27:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-01 18:05:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d11d3467bfba98535d38f1fb5e6c85da427a71e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-11-30 22:18:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba2ff80f9a6f5833df3373126cec4d8b9475f3c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-30 23:16:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cee123ec9494cc39cb7cd2482b9d462 : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bf436058c7f4eedbe6504e40eb2b64e : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e365f4fa9bbd4efb8f758b153374a9cd : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9125e96c9fcd4d4db95a831202ba5d53 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ebeb237bce9c4a95b554f4b358113490 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f60b05f7f8ac45f3b6f440b55bd7db37 : SUCCESS in 3m 25s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/e4aecc0fc6de444287545dbcc8ba3e7d : SUCCESS in 44m 23s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/548e140fba5a4d6a867866ddc1450f73 : SUCCESS in 47m 54s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/7a2f9c471d934b5cbe766093bcef410b : SUCCESS in 56m 24s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/9ab9411e858c4ad4bf2c6e6a4495b70f : SUCCESS in 32m 41s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/dc4d6d26283c406288c7d85feb62a551 : SUCCESS in 42m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ea3de2eafc45c67f9ff6c87dfcd6064335f4f24","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-14 18:32:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b3e6fe30f528e32410863a41892359fec903e8ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 18:53:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a38819ea4651437192656fb5353fd7d7 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6911cb994daa45b1b60820060e105960 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9766c96ebc8c44688ffb8d0bf0056a3d : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dffacebf088b4b76ab6d4330e0412029 : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b814d10a7da44eaa3632fe411a8feda : FAILURE in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/944f0fb4033d4465878f9a5ba35aed12 : SUCCESS in 5m 18s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/86691cad66a04e688264dbc000526690 : FAILURE in 10m 08s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/1357fc542d4a4f2c948010efcdbfd721 : FAILURE in 9m 16s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/7cebdfcbf0e54554ac82f2f675f6e398 : FAILURE in 15m 54s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/a2f33e394b1a40988adbf32e0fa8934c : FAILURE in 11m 23s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/528d7c3c83eb434a999fc2a96a830042 : FAILURE in 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"842d672aac463f531d3d6a9689d290e07a8b676b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-15 15:52:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f7f9090a971e49eb9903365924ce5036c667fb5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 16:08:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2588ffb92e534d6e81f3bb556a31865c : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dec3767541e243a8af20db7cd9de7a41 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515460a81add46ab9b6da965e9791fb5 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a50796ca1ee54a24b77404115d42e355 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9dd4d2c7aad4b829046ed7447e21365 : FAILURE in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/530a15dae1314c89aa182de7c4ceb323 : SUCCESS in 5m 43s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7d39e88c433b4382b93fc6b7f130546c : FAILURE in 11m 38s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/8f72bd9fd9144131a153609d4941cceb : FAILURE in 11m 26s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/a61c20436a7a4eeba14dfbd88713372c : FAILURE in 14m 04s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/912f90d32e1c4f20b9476216d1b6e068 : FAILURE in 11m 16s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/73f5588d5c0442268f169196183bc975 : FAILURE in 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d4478e9974f13dce50896d25d9c520df3a3d000","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-16 03:31:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"171252e81f801b08048d6f67382bb8d006893ccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 04:38:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13d269fead0f4467821d9aea9d70b62d : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31064a80311d4672b7eb3b456593b925 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a494ac119c7544e3a62a42f4b70d406e : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5a129da5ef1431cbe80e3018c9fbb4b : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04e2d76f84524aaf9205ec3a96f09e10 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2c17d177a304563b9456ffe4c47cb50 : SUCCESS in 5m 08s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/3c6f43ceeac5420997d39c867408af8e : SUCCESS in 47m 15s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/f22c4f8ee22a4410a2e049b2b567c75c : SUCCESS in 52m 07s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2a5861489e814381b6ed0c791ecb12b2 : SUCCESS in 1h 01m 42s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/34b53804c04d4b0e838c0a76b3ee8646 : SUCCESS in 38m 30s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/5b9f04485d6e4cfa8932f59461e2a4d7 : SUCCESS in 41m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39f47f0231007a48e386d69765cb183878b93338","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-30 17:27:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b2bd81cdfbc899cbef9adadb38bf06b5e6b71cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 17:27:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87c1e5e28a2a3b9871bee2f752680d1d7f1f6f9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 18:42:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcb3d6d10fc54ff4bdf858b825b2553d : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ef10cf311a14f2889d4d5ebd8fee92c : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c93a72fb06d48f0a91556e4b3bf4785 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d02798c1eedc4e3a843598afdfaef9a2 : SUCCESS in 16m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87dd24bdcb034fee85a1ea06f05138ba : SUCCESS in 7m 36s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9000f90817104f99ab6c51c35724da87 : SUCCESS in 1h 00m 20s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/1e638e0382e74c3e90a4e5af44c12d19 : SUCCESS in 50m 59s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/627026c4ffd9434ba6046e2b49a1138d : FAILURE in 1h 12m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e7368b1564bc8f01ac3f4e43e63d5457f6741456","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-31 02:05:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"19f03d8d55f68b6695636178f794d3573d2496e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 03:42:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cf17712030c4a00a5ace18040dda8bd : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/789c2beefabb4dc19d4700fb2c42d9f2 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c94168ba5281471c894b4f4c089588b2 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60e372a57f6845a884eae42dd830b59f : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a15a42b83b2450ebf51bb1b45bf7919 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5601f52a3def4dbf96f44e4d77152cc4 : SUCCESS in 5m 12s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/32187b5372a9461d9fde3b20ee89c5ed : SUCCESS in 52m 13s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/716bb327e34646ca8842ee56285a7547 : SUCCESS in 49m 57s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b58e4c6b4e724f4aa0d2f8b328377ffd : POST_FAILURE in 1h 35m 52s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/b53cd095605843818888406811292901 : SUCCESS in 35m 59s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/71255867305c4e0baa1270a767dced90 : SUCCESS in 44m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"305dee8cc7bf24473b8c8f2b586231c9b6ca3c19","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-01 15:43:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"adf7027703770a8f567f6496626ab5ef2fa6dcaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-01 16:52:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f745e3c27cf64e3d84dcfdd2a2969655 : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34bb2488ebd1429696c71b4d1a46acec : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69433b595ba2443e838a0e2a673bc9f0 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34e4ce7fea864ebaa1587554d8338901 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7c14da202454941ad3ce96d3b3ccb3d : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6816a52752d347ae981a767e8967e622 : SUCCESS in 5m 37s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2c2517bada3d4f35ac91b04df8de55d6 : SUCCESS in 45m 57s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/7ef4d032e95c485bb0533c5d8bd13d34 : SUCCESS in 46m 50s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/aa05e5b223ef491ca4c680f520adc941 : SUCCESS in 56m 53s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/77218e758a7f4e18be62262f91ca00d1 : SUCCESS in 34m 02s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/6f1fce25f6954b969a7b57624b0915ea : SUCCESS in 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf071e1caadb72c7b730dc6ce94cc6736b3be7c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-01 16:53:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9cc04bb367aacd8dcc6c14a5b69867efd904b40a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-01 18:05:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ab1eb1109ab1bfe421d736796fe1a308875d3a68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-01 18:05:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d87578f84fd44959d39e7fec004fca6 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3eb0877d3444578241b0ebe4dda2d5 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d159ee397434e1b97a03912c40a39b7 : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/216a18659c01449a976a37312be3eaa7 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22482352c0704b59af9bd787d85cd7c2 : SUCCESS in 8m 58s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ad31fb1159ee4aee980208a9e661e32f : SUCCESS in 51m 24s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/469a08cef20e4d6d972fe7ea4e4c0ade : SUCCESS in 52m 22s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/4765f8d8353241408bf5206c34dea937 : SUCCESS in 1h 06m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d25686d08dcdab70bf2caa4693ee97dfc6f33332","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-01 18:06:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5edd5cefefd4bca80bb0f99aded02a8 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bcf9607516067877684e5d78de5a9983fe89a374","revisions":{"bcf9607516067877684e5d78de5a9983fe89a374":{"kind":"REWORK","_number":1,"created":"2019-11-30 22:18:58.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/80/696780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/80/696780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/80/696780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/80/696780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/80/696780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/80/696780/1"}}},"commit":{"parents":[{"commit":"803201b15607f0f7d80805030c9e31fcfac22bbc","subject":"Merge \"capsule: ignore image_pull_policy if disallowed by policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/803201b15607f0f7d80805030c9e31fcfac22bbc"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-11-30 22:13:44.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-11-30 22:13:44.000000000","tz":0},"subject":"Change default of resume_container_state as True","message":"Change default of resume_container_state as True\n\nThis config determine whether zun-compute will restart containers\nthat was running in before. However, this might conflict with\nrestart policy of individual containers.\nFor example, if users create a container with auto-restart disabled,\nsetting resume_container_state to true would negate this effect.\n\nChange-Id: Ie70ed401dea63e9397d98c82b05059212e4e31c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/bcf9607516067877684e5d78de5a9983fe89a374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/bcf9607516067877684e5d78de5a9983fe89a374"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
