)]}'
{"id":"openstack%2Fzun~700282","triplet_id":"openstack%2Fzun~master~Ia515de0698bbdabfd8e6a43c4da230869294b008","project":"openstack/zun","branch":"master","topic":"bp/add-annotations-to-capsule","hashtags":[],"change_id":"Ia515de0698bbdabfd8e6a43c4da230869294b008","subject":"Add \u0027annotations\u0027 to db layer","status":"MERGED","created":"2019-12-21 19:59:05.000000000","updated":"2019-12-22 03:13:30.000000000","submitted":"2019-12-22 03:12:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700282-1576984325644-af565c7f","meta_rev_id":"63c91c3f71fba6847983e5bd7565170420b24233","_number":700282,"virtual_id_number":700282,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-22 03:12:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-12-22 01:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-12-22 01:58:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-22 03:12:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e0c0daab8f25ec5ade70cee1504f8db763405da","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-21 19:59:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b68482cae56e996394f9720591601e72311ae8c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 21:11:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9649ade8f0b74e0ab7e3336ab4bb3da8 : FAILURE in 4m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7abb92d1ab845dc97c96e8c5603ba47 : FAILURE in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f722bd1647b241958eb38fd5e45baa28 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bad861cca68640a2a9bfbb6040e24536 : FAILURE in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fe2d7a0ade14980953bcfe086954592 : FAILURE in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3daa9d72591445f3a061b91ba681e7ec : SUCCESS in 5m 10s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/f4ada167daea426f905d58adb03e45db : SUCCESS in 45m 41s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/3fa4c7b0791c4ee7804b07da31e6ecab : SUCCESS in 45m 40s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/61eab2c2f7014e9a8ad3c6b2dee6cd45 : SUCCESS in 1h 06m 07s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/13ad8f58a5e2455c89e20f90f85fefda : SUCCESS in 33m 42s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/ce831f73745144f584273046490740bf : SUCCESS in 40m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd4e3e162c91f8f0355e1a10836358a2f23138e0","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-21 23:01:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec36ecc37e6b3abd5ab71ed3bd4a5a18a4f05c61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 00:22:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a4611c6dbda450587c620c26a0b9fbc : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/232b1838e9604687a6daf55913599836 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78bd449127d942efb76e5e28d62c4349 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2c89f822b7d41c18275af093b591018 : SUCCESS in 9m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/564ce5047aa74af3ad41ffd59f54a841 : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4816a05d9a6c4a94beb18a2f7399a53a : SUCCESS in 8m 50s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/0b10bb40c3524bce989a89ffaf94569e : SUCCESS in 57m 15s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/c5a7f470cf7648178648a07bba53ed87 : SUCCESS in 59m 37s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/0e3cb62893be43f49ebc19937727ba6e : SUCCESS in 1h 13m 07s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/c9d1f187ef1744788631354d625f1c52 : SUCCESS in 30m 20s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/0e65236007994bfd912de2bbd76b9ac3 : SUCCESS in 46m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7b42743c3c489b7aceff2f84f047164a066513d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-22 01:58:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e0677de889c3d76765d72041044325ec2d8239ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 01:58:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e6829ddb6f450c5575ce48b64bbb39ab875e32f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 03:12:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af8314a3c0fe48b0be0ebb8d585c4fff : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e47de35b3a5149bbb3ed2c9db62ed4dc : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b162ed9dd94d4ddfbc9a0729717bb982 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d2e40cd3b4240e882de9ba176cdec90 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25cd87c76de6477e833ba67240d8ad91 : SUCCESS in 5m 12s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/fa36a9a69e994de9b69de41669351414 : SUCCESS in 45m 42s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/3caa2446c12c45718d91b41878ca7f9e : SUCCESS in 50m 42s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/83fa09fd83394bbe8bca6ee1f002fbdc : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1d6579babf1924f1f431108ff50f498f9d064856","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 03:12:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a507564af21d1f13b7596885d91a9d5a6a8b9165","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 03:13:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f144f025e5441f18d0ab57b4bb033b6 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4a8e0301de6b45737d12a9f3488957b7b083267f","revisions":{"46031362a2c1ca5a1479b740a7e2c7f4f4f36c81":{"kind":"REWORK","_number":1,"created":"2019-12-21 19:59:05.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/82/700282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/82/700282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/700282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/700282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/700282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/82/700282/1"}}},"commit":{"parents":[{"commit":"e9e5473d15187c947de23f53b5b11d53aeae19e7","subject":"PDF documentation build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e9e5473d15187c947de23f53b5b11d53aeae19e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-12-21 19:58:28.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-12-21 19:58:28.000000000","tz":0},"subject":"Add \u0027annotations\u0027 to db layer","message":"Add \u0027annotations\u0027 to db layer\n\nChange-Id: Ia515de0698bbdabfd8e6a43c4da230869294b008\nImplements: blueprint add-annotations-to-capsule\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/46031362a2c1ca5a1479b740a7e2c7f4f4f36c81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/46031362a2c1ca5a1479b740a7e2c7f4f4f36c81"}]},"branch":"refs/heads/master"},"4a8e0301de6b45737d12a9f3488957b7b083267f":{"kind":"REWORK","_number":2,"created":"2019-12-21 23:01:22.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/82/700282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/82/700282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/700282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/700282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/700282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/82/700282/2"}}},"commit":{"parents":[{"commit":"e9e5473d15187c947de23f53b5b11d53aeae19e7","subject":"PDF documentation build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e9e5473d15187c947de23f53b5b11d53aeae19e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-12-21 19:58:28.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-12-21 23:00:48.000000000","tz":0},"subject":"Add \u0027annotations\u0027 to db layer","message":"Add \u0027annotations\u0027 to db layer\n\nChange-Id: Ia515de0698bbdabfd8e6a43c4da230869294b008\nImplements: blueprint add-annotations-to-capsule\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4a8e0301de6b45737d12a9f3488957b7b083267f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4a8e0301de6b45737d12a9f3488957b7b083267f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
