)]}'
{"id":"openstack%2Fzun~700730","triplet_id":"openstack%2Fzun~master~I3fe1d8a6c961745801b07612b8982940aa683b3a","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I3fe1d8a6c961745801b07612b8982940aa683b3a","subject":"Rename Network object to ZunNetwork","status":"MERGED","created":"2019-12-28 19:08:11.000000000","updated":"2019-12-29 01:13:03.000000000","submitted":"2019-12-29 01:11:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700730-1577581897933-69f755c7","meta_rev_id":"4b367e701a1f9d0405e44df82c762800ed0d0ffa","_number":700730,"virtual_id_number":700730,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-29 01:11:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-12-29 00:09:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-12-29 00:09:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-29 01:11:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfef52efe0eaeb58759daaac58dfa52f009b0317","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-28 19:08:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8a8887dca9bb8b8bf8eb8a03ab4fc014ae82d9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-28 20:11:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b72e94b1e984e72b483809b00eaa3fc : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e5f03cd3aa44135a97d1acc693de910 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba5bf95305e4d15912579cf91797907 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee50e008c22b4a6cbccf5bbb4c448e70 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a00e1248a72c4c90abb36d2c8f1190b5 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b07e4b8f8714ac2a8633a22bcb03cec : SUCCESS in 7m 08s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/78b7dce05ede4a9986d78b425acdc1d3 : SUCCESS in 44m 21s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/fbd7a14225f54e478c5524131464a1d5 : SUCCESS in 42m 00s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/376c638e226f47e3b76b7344b5d1df61 : SUCCESS in 56m 37s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/d5b271baaef144f6a8679bac63f0cd3d : SUCCESS in 35m 05s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/160377c900ab41c19c58b7629a69671f : SUCCESS in 40m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf112a5743d17f50ffc42a04a2dd49988919bff2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-29 00:09:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1367eff8c7283acb37ccd6f934a69f9a7b46d4c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-29 00:09:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d22cfc92dcde5eccccf2a6207b609b2e8ee25b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-29 01:11:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0366071daa7946ebbd725b170cf5223d : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3297a99659f4823b277a81b97f749fd : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff0c07157e3b40aeb65fffa2719e4447 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bd39c77bdd9477ea5d385ad9ec213f6 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/162b02aac6dc41d88e7fc22c9f2e2231 : SUCCESS in 6m 09s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/128c22821f7a42f08fb243a8b6a76087 : SUCCESS in 47m 19s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/eaead9f072ed402fb1b06805df46bee8 : SUCCESS in 46m 11s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f690d92afac543fb852beb54bcdc4fd7 : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1150804bcea81620bb37d6073bd38a5928b07b4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-29 01:11:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4b367e701a1f9d0405e44df82c762800ed0d0ffa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-29 01:13:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8270b45cc12e48b0ae28392134c7f92c : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29df26207566717f0d1009091d25ea63b3b1d771","revisions":{"29df26207566717f0d1009091d25ea63b3b1d771":{"kind":"REWORK","_number":1,"created":"2019-12-28 19:08:11.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/30/700730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/30/700730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/30/700730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/30/700730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/30/700730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/30/700730/1"}}},"commit":{"parents":[{"commit":"4a8e0301de6b45737d12a9f3488957b7b083267f","subject":"Add \u0027annotations\u0027 to db layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4a8e0301de6b45737d12a9f3488957b7b083267f"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-12-28 19:06:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-12-28 19:06:45.000000000","tz":0},"subject":"Rename Network object to ZunNetwork","message":"Rename Network object to ZunNetwork\n\nThere is a Network object defined in os_vif library. To avoid\nname conflicting, we rename the object in Zun.\n\nChange-Id: I3fe1d8a6c961745801b07612b8982940aa683b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/29df26207566717f0d1009091d25ea63b3b1d771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/29df26207566717f0d1009091d25ea63b3b1d771"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
