)]}'
{"id":"openstack%2Fzun~700781","triplet_id":"openstack%2Fzun~master~I6e57b4f5ff7d63a4f876de6453f1aa8c8ca95d23","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I6e57b4f5ff7d63a4f876de6453f1aa8c8ca95d23","subject":"Remove unused db column \u0027meta\u0027 from \u0027container\u0027 table","status":"MERGED","created":"2019-12-30 17:20:56.000000000","updated":"2019-12-31 01:39:47.000000000","submitted":"2019-12-31 01:38:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700781-1577756298735-a23e576b","meta_rev_id":"f63467a6f6b1faabbefa17b92c635a59ffc18a4d","_number":700781,"virtual_id_number":700781,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-31 01:38:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-12-31 00:41:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-12-31 00:41:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-31 01:38:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c02ee31a9cfd51484c6efd2ea3694cfee1d05865","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-30 17:20:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3e74668473e31c88c5575fd25b4b2c6ddfb0c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 18:32:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99f0780e28e346fea6799ea4a9e0fd02 : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3ddf400d2a8463daafecaf265ea4bee : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13bd200e3e03480ba0460ab30c08edaf : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b73be409d1d4992850ee2412e938fa8 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/099316106a3345188fb5c99ae4fe771b : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b8bcc29dfa4490dbd807df006ed359d : SUCCESS in 5m 42s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/e9d70829399448dfbd629cb80bf5596c : SUCCESS in 55m 28s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/9b8b6ef0186c4e209439e60a91404fde : SUCCESS in 59m 55s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/fed399628e36401c9d3bf43a1bdf5264 : SUCCESS in 1h 10m 43s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/4c8d7a42dddd4e6ab0ccd0f30e06909d : SUCCESS in 41m 52s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/3e79e6bd3d434412a9a9bdf65ff1bb5b : SUCCESS in 51m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f522ba3b0a07d49523770a30ebf321a9e90de106","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-12-31 00:41:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"82289357f2c0f36b428dbccc03e3a64c87f0ba4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 00:41:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24ca4c2c3a940221864889dfcf38a9d60c5218f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 01:38:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c646a5dfbe0343cfb8cfac3b8fac031c : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23d9e9de24394008a1694c9ff7e10e7d : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32269c976d564d44ba67790efd4ee4d1 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d266db5cf88456d91b067b25a2d3cad : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f616c7cb14b548c69a60b16f5fdcd76d : SUCCESS in 5m 34s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/012e547b131647f58c27f6fb2878e957 : SUCCESS in 50m 22s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/4af9c76ad9e6414ebb954e5d4d65b91b : SUCCESS in 45m 16s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/57e93bb8708542e49e16c4ceeaaeb52a : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5994d9759934bcad90c5e1f561d51488a5a7c633","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 01:38:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f63467a6f6b1faabbefa17b92c635a59ffc18a4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 01:39:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd5bcfeb69cf4e1bba739f2903e0af69 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a6cbe7d05b8ac190823c2232e09d1dc70a9fa82","revisions":{"4a6cbe7d05b8ac190823c2232e09d1dc70a9fa82":{"kind":"REWORK","_number":1,"created":"2019-12-30 17:20:56.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/81/700781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/81/700781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/700781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/700781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/700781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/81/700781/1"}}},"commit":{"parents":[{"commit":"3f5178b7dc9520df84cfbf1de1e5156ce14d20c6","subject":"Merge \"check volume name in capsule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3f5178b7dc9520df84cfbf1de1e5156ce14d20c6"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-12-30 17:19:50.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-12-30 17:19:50.000000000","tz":0},"subject":"Remove unused db column \u0027meta\u0027 from \u0027container\u0027 table","message":"Remove unused db column \u0027meta\u0027 from \u0027container\u0027 table\n\nChange-Id: I6e57b4f5ff7d63a4f876de6453f1aa8c8ca95d23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4a6cbe7d05b8ac190823c2232e09d1dc70a9fa82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4a6cbe7d05b8ac190823c2232e09d1dc70a9fa82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
