)]}'
{"id":"openstack%2Fzun~703259","triplet_id":"openstack%2Fzun~master~I66fed92c4f67b4552aca3bd8d81b6dda32d70442","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I66fed92c4f67b4552aca3bd8d81b6dda32d70442","subject":"Load container driver from entry point","status":"MERGED","created":"2020-01-18 22:17:14.000000000","updated":"2020-01-25 21:26:59.000000000","submitted":"2020-01-25 21:25:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703259-1579987536615-17819b5d","meta_rev_id":"b945be1bb16b0141d45a4becff536e439d5c3d42","_number":703259,"virtual_id_number":703259,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-25 21:25:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-01-25 20:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-01-25 20:21:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-25 21:25:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21c3b60471e189a6db9a6f5299762a585a3a53e6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-18 22:17:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"651b5062737229d287b23177245bf8ef86a96558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 23:40:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1cfac8b4a979477085d1795cba034285 : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bd178cebb7546dbb9d95054fffb3145 : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc6a339048e14cb3a1fa5e19a03fc46b : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96d7fe4f297d4fbd9f406805ccf5d4b8 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/988afad95c9241deb34b04e626fed7cf : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/864b1dc243f54dd0ad458a6aac715481 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32fed29f7f864e529a09e09aff8252ae : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5be8c52196e4a819226864c9afd7114 : SUCCESS in 4m 05s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9b4a0746d3e44e029e1c94e2ac956bb9 : SUCCESS in 48m 28s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/f92b81c2fa3047c0ab16d178edb78fa8 : SUCCESS in 48m 44s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/733d803498a34dd98f68eb2461cf6a16 : SUCCESS in 1h 20m 02s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/fd12b5adc13f4f759e018a4992d3db65 : SUCCESS in 48m 14s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/64bdbb7779d94faa931bd36a6c46306e : SUCCESS in 53m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07fa75d874394c84380e8fc8744b8b548369b8a1","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-19 23:56:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6a28a1814af821d3df8c433a2e4294218d3544db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 01:09:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0a73ed0b8a6344ccb4b73294e77d0f3d : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0509acccaf0d420889dda8654879928e : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a7fb3edfafb41af8bda08db66d05489 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33d8960618344791876f486a16dd02ae : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab028193a6f645bbab1020eff2ab6ac6 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01689c8741aa43249e015472b71de40d : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bff777e7ef4e490fbbd8132052e61f9b : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc468ccbaea94e93a39fb0a8ac6b56c1 : SUCCESS in 4m 10s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a5d99f9e279a4efab7ef951af352b891 : SUCCESS in 52m 01s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/b7dc4569211b4df3a157e4713578eb7c : SUCCESS in 56m 02s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/458d7574940f49f2bee9b5e393216964 : SUCCESS in 1h 10m 19s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/00ad2c232de34888897e2ca736e6f87c : SUCCESS in 39m 04s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/134c62ea4ece442c978279d4d7f79b4b : SUCCESS in 43m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2eb9aabf92ca253ba019707a8aea1948567ba183","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-25 19:09:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f8583c9652001c13fb84c05e5f0b2dc8a5db6f9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 20:12:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/852ecdf9003f4030bd5ccb164ea8871c : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65c325adc367476bb8fc46de9d2e4497 : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df75e905e45d4a70a7f9a7f5091ef9aa : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f614f8bc2784c89b3bdc9b6a1574305 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2181ace9501c4c0e9ce1ab15dc72ec3d : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03b0122b8ab44bdb97bd7fbf83908298 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd8eb0c260b84dfea1280e757322f4d3 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d17408738d415b825acec09cc2fcfb : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ff763cff99947798fa0c4d33a039f52 : SUCCESS in 4m 31s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/00afd20833a44907abe19063c3792916 : SUCCESS in 1h 01m 23s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/9f1aa107bcce448fa6a19d301e9acce6 : SUCCESS in 45m 55s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/5a2a3069758e4fab91f338b73856703a : SUCCESS in 57m 45s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/573cb25a8b264038ab3afb4c7fe0b709 : SUCCESS in 31m 49s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/dbac1986a359439db0b309b56fd5db8c : SUCCESS in 38m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4bfc8308997b7392dee094113b273c5530b55017","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-25 20:21:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0553f31c7c78340736f01e67df86cf20ce9349f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 20:21:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2e5f7e6dd9bc7fba9c21950bd9b7d1918b893619","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 21:25:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6b342f0bcbbf48b4b7f64d96317830dd1ba36e0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 21:25:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f7465349b7fb41d9aba80bd4e65cee68 : SUCCESS in 2m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f60ab6450c6545f7bb0a82413eb5d5f9 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c84f03d855574cdb953f7ec9f149ad81 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1cfa762c4c0478aa8731c55242819b9 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b2d8470f6ff4d7db54513afb97f9e2d : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74751bfeff18425f8cb12807caebbff3 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/809474ae2d824c65b8c90ed119ef16e6 : SUCCESS in 3m 30s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ab001684a6054aad8d09835d12e7ce4e : SUCCESS in 44m 56s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/668ecd1f476f49a6a25da292d0b4e691 : SUCCESS in 47m 21s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f8e5021e74c34a1b863cb4c48c14993a : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":3},{"id":"b945be1bb16b0141d45a4becff536e439d5c3d42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 21:26:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f446a048e246e89a89b9353bd99be6 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58b545db45b547b1a5f1e28a2b317b79 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9f7922a7ca2eb3e9ee1c15fcb78fc2706fffb280","revisions":{"ec017135cdcd4aa2d9144476ea174bbd363e2c36":{"kind":"REWORK","_number":1,"created":"2020-01-18 22:17:14.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/59/703259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/59/703259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/59/703259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/59/703259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/59/703259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/59/703259/1"}}},"commit":{"parents":[{"commit":"519f7acde422001111a258c8aed28c741a621f85","subject":"Move neutron binding constants to zun/common/consts.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/519f7acde422001111a258c8aed28c741a621f85"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-18 22:16:59.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-18 22:16:59.000000000","tz":0},"subject":"Load container driver from entry point","message":"Load container driver from entry point\n\nChange-Id: I66fed92c4f67b4552aca3bd8d81b6dda32d70442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ec017135cdcd4aa2d9144476ea174bbd363e2c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ec017135cdcd4aa2d9144476ea174bbd363e2c36"}]},"branch":"refs/heads/master"},"2d9cfe8f93b757941bc2e5e3356241e512bef907":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-19 23:56:22.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/59/703259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/59/703259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/59/703259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/59/703259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/59/703259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/59/703259/2"}}},"commit":{"parents":[{"commit":"b0f8488d952150d07cc7489461783279823eb907","subject":"Move neutron binding constants to zun/common/consts.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b0f8488d952150d07cc7489461783279823eb907"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-18 22:16:59.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-19 23:54:53.000000000","tz":0},"subject":"Load container driver from entry point","message":"Load container driver from entry point\n\nChange-Id: I66fed92c4f67b4552aca3bd8d81b6dda32d70442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2d9cfe8f93b757941bc2e5e3356241e512bef907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2d9cfe8f93b757941bc2e5e3356241e512bef907"}]},"branch":"refs/heads/master"},"9f7922a7ca2eb3e9ee1c15fcb78fc2706fffb280":{"kind":"REWORK","_number":3,"created":"2020-01-25 19:09:15.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/59/703259/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/59/703259/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/59/703259/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/59/703259/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/59/703259/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/59/703259/3"}}},"commit":{"parents":[{"commit":"b0f8488d952150d07cc7489461783279823eb907","subject":"Move neutron binding constants to zun/common/consts.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b0f8488d952150d07cc7489461783279823eb907"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-18 22:16:59.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-25 19:09:09.000000000","tz":0},"subject":"Load container driver from entry point","message":"Load container driver from entry point\n\nChange-Id: I66fed92c4f67b4552aca3bd8d81b6dda32d70442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9f7922a7ca2eb3e9ee1c15fcb78fc2706fffb280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9f7922a7ca2eb3e9ee1c15fcb78fc2706fffb280"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
