)]}'
{"id":"openstack%2Fzun~703301","triplet_id":"openstack%2Fzun~master~I4a68b58f46bc7eaac494909647b87ada7813a8eb","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I4a68b58f46bc7eaac494909647b87ada7813a8eb","subject":"Refactor get_available_resources in container driver","status":"MERGED","created":"2020-01-19 18:49:12.000000000","updated":"2020-01-26 16:50:10.000000000","submitted":"2020-01-26 16:48:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703301-1580057327151-17819b5d","meta_rev_id":"7726d36838c34f199ce2045e21d91a3fa73b45eb","_number":703301,"virtual_id_number":703301,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-26 16:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-01-26 15:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-01-26 15:43:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-26 16:48:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"800c226cf0d8af2a06971ea9ca1573d97a09ff43","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-19 18:49:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8aeb5b15e39c3b6354df3b2a4066baea629348c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 20:09:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe3fe3cc459742dc845aa2a11aace61f : SUCCESS in 9m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6298d95c6be34664ac5dc5cb5ea14c30 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/414e092f7e5f4142b2852ae4508eed6f : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb2f8f4bbbfe49cf8b2693c9e260ceca : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14e550bb79dd420a92dbed5743e35415 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c36d641270364cf3895482020222c27d : SUCCESS in 6m 44s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/71d360a4069b41d79e3e612b22420671 : SUCCESS in 39m 20s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/ea2d5eb825064e7abe8d64550065c684 : SUCCESS in 51m 21s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/84b01d136ee649c7a5568b88e0966c43 : SUCCESS in 1h 05m 43s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/84d379f7544b45a0aa88510d10b1a173 : SUCCESS in 33m 25s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/cd4e7f776a524154add3a17dc871a1fb : SUCCESS in 41m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e631e0f3f81f22f5b16a151915aba716c03c8cd","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-19 23:56:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8b391d41282106298ec12cb12ce343e97da6dbd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 01:03:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7c0eef3b8624fa89c0a361998f09904 : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dafe84143e44d42977a7c1d54b43de3 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95c1c5ecdc15432b9396de810d8092cd : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de6c9576c26548ffbeac1221b2f82bcc : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41e3ffbeae43499682fbcdb47716b288 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc5beede90749488a1ab12a4bea4ac5 : SUCCESS in 5m 45s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/207ea2d8ad5c41069203210d0ea113d0 : SUCCESS in 48m 25s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/01b4c291fa2f41399df16d16352c870b : SUCCESS in 48m 56s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/de73ec473fbc4ab1975a5931286eb112 : SUCCESS in 1h 00m 48s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/a2db55d8cbcb4c95942e1a7352e93134 : SUCCESS in 38m 17s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/b5c72bf9aa224dda95429e18e7a84f8a : SUCCESS in 46m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b841b6def2788890655ddcf4abae2023b3c3127b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-25 19:13:26.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7aff5177cbd43e0cfc74f1757485bf6e86d56eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 20:15:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a088f3cab5c54aba8afbcc2e22916320 : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72f486bc3ddc41039c13cab0a47733e8 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fce5b2053ce4fe5a31079bb14fce574 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/578ac5b751ff4dcaa7d9d40feddb1ea0 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1d183bf74f748b68f4426fcd2c622cf : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c325993cc3004a55aaca1bd4dd0f1999 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/604fc9fe11204f9889f585eefe6063c6 : SUCCESS in 5m 43s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/8fc8a4d3d71e45c89e996d75ecb5d6d1 : SUCCESS in 46m 50s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/0d1e5a49dd57419d9ae8d1b32554ccd3 : SUCCESS in 47m 04s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d1556ce3d33e48e1b1f10f6f39832eb4 : SUCCESS in 58m 23s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/a83e2fbb62ea442991c4295e5d2f6e7c : SUCCESS in 32m 01s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/b3cd408c69914240a2c3edc0acbd1d05 : SUCCESS in 42m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3230a5f11afa268e99f7e466a98139073a876e92","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-26 15:43:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a317c00b1655613245c9889e75446fb842145c0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 15:43:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"328de676fd6f17e2a1971ddfdacac8409bac6fd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 16:48:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a296be8817284a27a3645aca588b410a : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1eef594cac34300b430595378198115 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1473a629c04342c5907083e587842ea2 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04f01f9347ab4bfd891757ac946b8e19 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47a059b443ca448697ea761def218071 : SUCCESS in 5m 15s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/fbb65160436f47d89294632853fa8a3f : SUCCESS in 48m 10s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/61ff4f790054479897c5b25df443483f : SUCCESS in 44m 03s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b2575a623dcd4121a6816a82d1e0dcea : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":3},{"id":"148cee04c3ec4d2bd46fb630975c65342fe67fd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 16:48:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7726d36838c34f199ce2045e21d91a3fa73b45eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 16:50:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d27ef870f234e8b9a4d0295a06be1e6 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"428f9f83751b4040f27f1b660672bfe1e1451e00","revisions":{"de4452dfedfa95d32b8eb8185b742f942513c764":{"kind":"REWORK","_number":1,"created":"2020-01-19 18:49:12.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/01/703301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/01/703301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/703301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/703301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/703301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/01/703301/1"}}},"commit":{"parents":[{"commit":"fd947d0c2fc660c8a12a3bee33ddead690671c73","subject":"Provide default implementation for several methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/fd947d0c2fc660c8a12a3bee33ddead690671c73"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-19 18:36:13.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-19 18:41:22.000000000","tz":0},"subject":"Refactor get_available_resources in container driver","message":"Refactor get_available_resources in container driver\n\nMove docker-specific code to DockerDriver subclass.\nAfter the refactoring, generic code stays in base class and\ndocker-specific code is moved to docker driver subclass.\nAs a result, we can re-use the generic code in base class\nif we introduce a non-docker driver in the future.\n\nChange-Id: I4a68b58f46bc7eaac494909647b87ada7813a8eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/de4452dfedfa95d32b8eb8185b742f942513c764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/de4452dfedfa95d32b8eb8185b742f942513c764"}]},"branch":"refs/heads/master"},"637f9f4d8934a1e8a760125692698c17e3ba5251":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-19 23:56:22.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/01/703301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/01/703301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/703301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/703301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/703301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/01/703301/2"}}},"commit":{"parents":[{"commit":"475ee5c624382edfa0d3be5bc6c622bca8254f84","subject":"Provide default implementation for several methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/475ee5c624382edfa0d3be5bc6c622bca8254f84"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-19 18:36:13.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-19 23:54:54.000000000","tz":0},"subject":"Refactor get_available_resources in container driver","message":"Refactor get_available_resources in container driver\n\nMove docker-specific code to DockerDriver subclass.\nAfter the refactoring, generic code stays in base class and\ndocker-specific code is moved to docker driver subclass.\nAs a result, we can re-use the generic code in base class\nif we introduce a non-docker driver in the future.\n\nChange-Id: I4a68b58f46bc7eaac494909647b87ada7813a8eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/637f9f4d8934a1e8a760125692698c17e3ba5251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/637f9f4d8934a1e8a760125692698c17e3ba5251"}]},"branch":"refs/heads/master"},"428f9f83751b4040f27f1b660672bfe1e1451e00":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-25 19:13:26.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/01/703301/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/01/703301/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/703301/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/703301/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/703301/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/01/703301/3"}}},"commit":{"parents":[{"commit":"112c5a6cd6c0fa7ef1affb771620c1ba20cd2a96","subject":"Provide default implementation for several methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/112c5a6cd6c0fa7ef1affb771620c1ba20cd2a96"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-19 18:36:13.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-25 19:10:25.000000000","tz":0},"subject":"Refactor get_available_resources in container driver","message":"Refactor get_available_resources in container driver\n\nMove docker-specific code to DockerDriver subclass.\nAfter the refactoring, generic code stays in base class and\ndocker-specific code is moved to docker driver subclass.\nAs a result, we can re-use the generic code in base class\nif we introduce a non-docker driver in the future.\n\nChange-Id: I4a68b58f46bc7eaac494909647b87ada7813a8eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/428f9f83751b4040f27f1b660672bfe1e1451e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/428f9f83751b4040f27f1b660672bfe1e1451e00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
