)]}'
{"id":"openstack%2Fzun~704237","triplet_id":"openstack%2Fzun~master~I213fa67057a5fceaa9a150ab7c68aedd5f2eb4b0","project":"openstack/zun","branch":"master","topic":"bp/add-support-cri-runtime","hashtags":[],"change_id":"I213fa67057a5fceaa9a150ab7c68aedd5f2eb4b0","subject":"Move port deletion code to zun/network/neutron.py","status":"MERGED","created":"2020-01-26 00:03:36.000000000","updated":"2020-02-08 22:18:38.000000000","submitted":"2020-02-08 22:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704237-1581200211041-5f1f21a0","meta_rev_id":"d03f632cd58a34ef95657158859b365b05b901ca","_number":704237,"virtual_id_number":704237,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-08 22:16:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-02-08 21:15:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-02-08 21:15:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-08 22:16:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3e1850a69e2fff55fd4c2abf3b2731620a94817","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-26 00:03:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fb85f97f9ef9b9249faa8bddc73683d0c84dd5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 01:06:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaccfddd05af480790bffa728fdfa934 : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53ea9f72f63e415fa49c9f9ccd908c36 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3824beb5705f401bb83ae902e065191e : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f04e66eeff894597a6de3d7d12fcc864 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9eaf7071f851462dbec6014ccff0aeb0 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53adb9c4275d45ec94bf9592dcd19f26 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7488cfe7d2e44755be61ea7ec50c6be8 : SUCCESS in 8m 40s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/173c1cc7078940068ae6c36be93e5e12 : SUCCESS in 47m 33s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/202397d7434a4dc6872e89ff129ea59c : SUCCESS in 46m 10s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/c406259694a345409cca1baa727f104e : SUCCESS in 57m 52s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/9d541e30ca004238a5a5d84c4c2b5aab : SUCCESS in 33m 30s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/2fa14875591e4e3db8c91fb63b496704 : SUCCESS in 40m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f3ee35867043938215c3c468a800ba790de085f","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-02 17:44:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"dee4e17614228988d30eb409d69ced333e5b0f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 18:48:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/832f49146fb844d8a4a7ebfac4520a96 : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96628b901cf04844a2bf9833af0f2898 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b5dd7b17fa42e4addcdbc0057eb2e4 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bbb3c3f132a4d5eb526065ddb61bf96 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/972bfc9f7cc040ef87387b915e5da4f0 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/912a4e9b9a104699befc7f10cce9f8e4 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb016d7563944f71b91e0dfce982be71 : SUCCESS in 6m 57s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9fdfa60e63134965a36fedfe06086191 : SUCCESS in 46m 06s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/5851c95e554842c3921fc0b8cfaed645 : SUCCESS in 47m 16s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/12ce7d7d0c3243648c56e15ad08405c1 : SUCCESS in 1h 02m 26s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/fa2356aeaca841d086a9ec317d06835a : SUCCESS in 34m 21s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/2c1848033bd94a3f9dbc5359d881cb66 : SUCCESS in 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d13bbe40faade7a170759b1efe0abfa4a77d07fb","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-08 21:15:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9e523d38f7b72fb8ef6cc82480fcb8a4cffa5c8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 21:15:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"031dff02e874c5869277a7c0dc1ffa8e4190e29b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 22:16:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/818e8592d8654e6dbe0a99976443aa9b : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f230b770caaa45e1bd70d31da40a39dc : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7148e1aa49341b4a2989cdc6fdaa68e : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed4587931ff34541841f54bb09725b10 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fa17b634f64bf69cc5de037deefa52 : SUCCESS in 5m 05s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/8cfcb0b719b94813a70cf0cf91aaa9bd : SUCCESS in 46m 02s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/fa0021ceffa74db699e2b6d681039b10 : SUCCESS in 44m 31s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ab35053abe2a42d4b5c10471ccee0618 : SUCCESS in 59m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c252a4e0b9964886dfcd96e4c4b172c557b646bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 22:16:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"91684701e4cf31d9b90484a0af1cefc9ac469ed8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 22:18:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc640b77906b41b09097610f7a826680 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"63735680c3011429dd3ab920be8be684d5e38bc5","revisions":{"e7c957d9e02842fa9f50d3bb7de83f4fd1c55f65":{"kind":"REWORK","_number":1,"created":"2020-01-26 00:03:36.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/37/704237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/37/704237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/704237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/704237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/704237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/37/704237/1"}}},"commit":{"parents":[{"commit":"54640a22f3cdf951ce01436ad1b6f5aceed68e91","subject":"Move create_or_update_port to zun/network/neutron.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/54640a22f3cdf951ce01436ad1b6f5aceed68e91"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-26 00:01:59.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-26 00:01:59.000000000","tz":0},"subject":"Move port deletion code to zun/network/neutron.py","message":"Move port deletion code to zun/network/neutron.py\n\nThis allows the neutron port cleanup logic to be reused by other\nruntimes.\n\nChange-Id: I213fa67057a5fceaa9a150ab7c68aedd5f2eb4b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e7c957d9e02842fa9f50d3bb7de83f4fd1c55f65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e7c957d9e02842fa9f50d3bb7de83f4fd1c55f65"}]},"branch":"refs/heads/master"},"63735680c3011429dd3ab920be8be684d5e38bc5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-02 17:44:42.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/37/704237/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/37/704237/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/704237/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/704237/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/704237/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/37/704237/2"}}},"commit":{"parents":[{"commit":"da8e5e4e9c330794e16368827260cd2b65e90f87","subject":"Merge \"Move create_or_update_port to zun/network/neutron.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/da8e5e4e9c330794e16368827260cd2b65e90f87"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-26 00:01:59.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-02 17:44:11.000000000","tz":0},"subject":"Move port deletion code to zun/network/neutron.py","message":"Move port deletion code to zun/network/neutron.py\n\nThis allows the neutron port cleanup logic to be reused by other\nruntimes.\n\nChange-Id: I213fa67057a5fceaa9a150ab7c68aedd5f2eb4b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/63735680c3011429dd3ab920be8be684d5e38bc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/63735680c3011429dd3ab920be8be684d5e38bc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
