)]}'
{"id":"openstack%2Fzun~704261","triplet_id":"openstack%2Fzun~master~I2afe6eba3cfd9687a0eea390f23e95a75888aaec","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I2afe6eba3cfd9687a0eea390f23e95a75888aaec","subject":"Start zun-cni-daemon in devstack","status":"MERGED","created":"2020-01-26 22:33:37.000000000","updated":"2020-01-28 02:46:10.000000000","submitted":"2020-01-28 02:44:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704261-1580179471239-a46f390a","meta_rev_id":"11f962b6695c6b37363b9f2cc076fcc6ee78a712","_number":704261,"virtual_id_number":704261,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 02:44:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-01-28 01:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-01-28 01:37:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 02:44:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd0d3b29804b7381872835443dd937ab27a5ba53","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-26 22:33:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37592349065f6ae17614e731e1c364f7cb323a0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 23:48:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4ce25d4dea3440bad7bcdbbf51ac31a : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ce03adeff134dbaa8f84cdc817d8ea1 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd0bd337d27f4cb8b2927d82849a6dd4 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82adcdff7dc44257b16ef2b75502d7c0 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dcd7e8f96e3d44f89efb8b64557be08c : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26449930e97f47b595f7cf17fb90cf5b : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d43112993445bcbe6d6f07929975b4 : SUCCESS in 4m 58s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/063c1f0b23614632bad3ac8bfa06ed42 : SUCCESS in 43m 59s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/c5363046f57d4b40be660b24cf38c8a2 : SUCCESS in 45m 19s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d20a15fae7a24500a7f6178b6bc4db96 : SUCCESS in 1h 08m 49s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/721b2770f5fd49dfb86d864a09cdf610 : SUCCESS in 33m 36s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/7aff2cff4b084fc99f333c1987af7c52 : SUCCESS in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3066c62f0cdbc60eb4e8b2da9688ddb17082c82b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-01-28 01:37:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b0a2e2ef102b01dacd850e95fe59463b4d7e793","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 01:37:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"937c84bf9fea4d5f5c4c30c6a94a3d05f76c754c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 02:44:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d23573eab7b8477f9142181086192333 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c9b368de4d444a91d536c8976d4714 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c76e3523ce2d4de6bb3b5e6c768a0c42 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/156553edecc84bdc8c85280b322e85e5 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed1cc9b06a4c423ab71c3904ffbe2618 : SUCCESS in 5m 22s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/00f5662551bd4b81a04552fead01ed1f : SUCCESS in 46m 07s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/b27b22ff6f82417d86a8f02fa41217d0 : SUCCESS in 49m 59s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/c44b9d07d3b14e8ea2053a3a4f756d52 : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3440214fad3ed04ded88cc43e18cd208c7af61a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 02:44:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"11f962b6695c6b37363b9f2cc076fcc6ee78a712","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 02:46:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee1c02b73b7646f983d929f9fb472798 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69844fe9c7108d5b1c2efccd4048e37cf1e9037d","revisions":{"69844fe9c7108d5b1c2efccd4048e37cf1e9037d":{"kind":"REWORK","_number":1,"created":"2020-01-26 22:33:37.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/61/704261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/61/704261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/61/704261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/61/704261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/61/704261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/61/704261/1"}}},"commit":{"parents":[{"commit":"428f9f83751b4040f27f1b660672bfe1e1451e00","subject":"Refactor get_available_resources in container driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/428f9f83751b4040f27f1b660672bfe1e1451e00"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-26 22:33:19.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-01-26 22:33:19.000000000","tz":0},"subject":"Start zun-cni-daemon in devstack","message":"Start zun-cni-daemon in devstack\n\nChange-Id: I2afe6eba3cfd9687a0eea390f23e95a75888aaec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/69844fe9c7108d5b1c2efccd4048e37cf1e9037d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/69844fe9c7108d5b1c2efccd4048e37cf1e9037d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
