)]}'
{"id":"openstack%2Fzun~706177","triplet_id":"openstack%2Fzun~stable%2Frocky~I6182926db0c314338c03ed02b1d2f28b057f0eb3","project":"openstack/zun","branch":"stable/rocky","hashtags":[],"change_id":"I6182926db0c314338c03ed02b1d2f28b057f0eb3","subject":"Replace tox_envlist \u0027all-plugin\u0027 with \u0027all\u0027","status":"MERGED","created":"2020-02-06 04:35:23.000000000","updated":"2020-02-09 16:45:50.000000000","submitted":"2020-02-09 16:45:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706177-1581266750223-85323f19","meta_rev_id":"68f8d1bf66fdaf6217ed150d5ea878ed9117bec5","_number":706177,"virtual_id_number":706177,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-09 16:45:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-02-09 15:55:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-02-09 15:55:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-09 16:45:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04f0145ab93d3031fc8c9e53cdd8ece7f412c827","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-06 04:35:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"b38f1a7a59c93a63420afb7f7a22cf1d75c84c81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 05:04:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe6917b1b16c4d7c9e732b077027b365 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7e51061eae741b88e1850007fc3c946 : SUCCESS in 5m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/259a47cfdc6e4f73a3b37181dcef2a08 : SUCCESS in 6m 29s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/66aa1eaa316e4d4fb5fe4b4607fafb16 : SUCCESS in 4m 30s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/e1d680dbd3aa4ce7bffe853f2b678dd1 : FAILURE in 18m 59s\n- zun-tempest-py35-docker-sql https://zuul.opendev.org/t/openstack/build/93bade8e11d542ef8624cd0ab1c5d56d : FAILURE in 21m 50s\n- zun-tempest-docker-etcd https://zuul.opendev.org/t/openstack/build/0c264df8728140e680362115922a4745 : FAILURE in 17m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/6b99bad8df1b49d387823be2b2bddb54 : FAILURE in 23m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe2467960a164ade957d84181c410154 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e1c1e13ab695ee90c01346f004f8413cc8e526b7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-08 17:04:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"408116d3c7a13365b7367812ef72893b9284381d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 17:22:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206005ea6216498ab8b06fc538728ce9 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbce1222d6634144b8d50a684f560ebb : SUCCESS in 4m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d76ec7e95c5b4893b9c44e340c992468 : SUCCESS in 4m 50s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b0b4971f872b4ab9b698184c5d81fe01 : SUCCESS in 6m 20s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/94b51a352fd64277b1b9b601cd744495 : FAILURE in 14m 01s\n- zun-tempest-py35-docker-sql https://zuul.opendev.org/t/openstack/build/7e7aa71c769340509d9f744441b53c1e : FAILURE in 14m 25s\n- zun-tempest-docker-etcd https://zuul.opendev.org/t/openstack/build/fe9d191e62f9461e8dc0f8c78e07d4fc : FAILURE in 14m 15s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/92f98710d908485d94ec680e0ebd53ae : FAILURE in 16m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eafead27e5194aa396eac0b2be725f2e : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d99561cb084106435b392ce8292496544c761fa1","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-09 00:07:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e1edd1709c809407b2133ef8c6f1a00dd8533fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 00:29:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a2ba88a33e4b6db46fe48a6e188cb5 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/661c53c506e249f5a040242b9818747c : SUCCESS in 5m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c73bc406a96b4ffa8b6504e6a5e9f27a : SUCCESS in 5m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1290e5c7a56441568f0a2768af15a1f7 : SUCCESS in 4m 58s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/afdf69de5e4d47aaa3caf73ba344d97d : FAILURE in 17m 14s\n- zun-tempest-py35-docker-sql https://zuul.opendev.org/t/openstack/build/62fa3f0cabca418f8d32f015b1ef93d3 : FAILURE in 17m 30s\n- zun-tempest-docker-etcd https://zuul.opendev.org/t/openstack/build/0bc3368772ab4030bc6f1dd263c235fc : FAILURE in 17m 04s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/1e95aef74fb3425cb950375357041aaa : FAILURE in 18m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19ad09ec39d849df824745f6eb906d3b : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"efd71aa48349e43f9db31d1fb077711b55a2ac2d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-09 14:08:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4a1818b69503db48af17a1d83e08e696b9949be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 15:33:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a146125ac444bbf83e2263bca31188f : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1deffa3a37024e6b816564511b46eda9 : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9374fc3e7193421b86e03f3d3f3ddcdf : SUCCESS in 5m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f7b17550c8254544a5c698dc47960aad : SUCCESS in 4m 32s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/2a6c4ad7a2384e09801ecb5e846d9012 : SUCCESS in 42m 41s\n- zun-tempest-py35-docker-sql https://zuul.opendev.org/t/openstack/build/a3603efa934c483183bf6b1d6d636bd1 : SUCCESS in 57m 16s\n- zun-tempest-docker-etcd https://zuul.opendev.org/t/openstack/build/46ba97bd547d4a7c935769460f3716e7 : SUCCESS in 41m 27s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b7e99adaf3344e9b9d21e1a210f7f771 : FAILURE in 1h 23m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b8deb95c02b4ed98b2a101e3540a574 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5e0aa143df791bfd32c1aba196ee43f6f7cd2b4b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-09 15:55:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e21ffba1d8ca3e9563a04ab92162bf5c6205f94e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 15:56:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"641d565dceb11d5f8c2c73faa763128051e5bcb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 16:45:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"68f8d1bf66fdaf6217ed150d5ea878ed9117bec5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 16:45:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9855cd721438440eade3662d88c5081b : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8fd460ba1e5646088f7bbe449e691362 : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/df74013977234203a18329e0561d9e15 : SUCCESS in 5m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/908ca23e712c4beb89d91fc3c1d7269d : SUCCESS in 4m 39s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/a394b1c9504049a8aa8e132550d5e9ea : SUCCESS in 43m 44s\n- zun-tempest-py35-docker-sql https://zuul.opendev.org/t/openstack/build/620f4f718c8d44b39c3fb6d3a1e79b2b : SUCCESS in 48m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86ef8d8ea96e4ca3898850961c055ed4 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1be72f5885b67155fbb923ec7a37615a773c828b","revisions":{"1be72f5885b67155fbb923ec7a37615a773c828b":{"kind":"REWORK","_number":1,"created":"2020-02-06 04:35:23.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/77/706177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/77/706177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/77/706177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/77/706177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/77/706177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/77/706177/1"}}},"commit":{"parents":[{"commit":"8ff2d04a90a9761268dcdb310c7992ce577a2608","subject":"Increase exec timeout to 15 seconds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8ff2d04a90a9761268dcdb310c7992ce577a2608"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-05 05:25:36.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-06 04:35:23.000000000","tz":0},"subject":"Replace tox_envlist \u0027all-plugin\u0027 with \u0027all\u0027","message":"Replace tox_envlist \u0027all-plugin\u0027 with \u0027all\u0027\n\nThe tox envlist \u0027all-plugin\u0027 is deprecated in tempest\n\nChange-Id: I6182926db0c314338c03ed02b1d2f28b057f0eb3\n(cherry picked from commit 67b1ec978ff4365fc0d00eb6fe66a796e0fe15cf)\n(cherry picked from commit 25e566361ee654d9ebfb086341ff5c20ac84823a)\n(cherry picked from commit 3f4ea79e2d9fa56b9d1ae3474be19ebab5a52d68)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1be72f5885b67155fbb923ec7a37615a773c828b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1be72f5885b67155fbb923ec7a37615a773c828b"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
