)]}'
{"id":"openstack%2Fzun~707526","triplet_id":"openstack%2Fzun~master~Ife71fb32e0a091aaef30e97e626be307ff851356","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ife71fb32e0a091aaef30e97e626be307ff851356","subject":"[WIP] support to attach/detach network to capsule","status":"NEW","created":"2020-02-13 04:47:06.000000000","updated":"2021-07-10 08:54:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9cc33aa328483b8995f2445605a4bd25d8917e8c","_number":707526,"virtual_id_number":707526,"owner":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-10 08:54:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 02:06:58.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-03-30 03:07:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b91ecc533e67c1b88ede9cba959ea793eae7f363","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2020-02-13 04:47:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cae67f0c0a9b955832dd958abac65cef15f10fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 05:55:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/307f42b5614f4e229f1645fd0c208bec : SUCCESS in 7m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9235d9ec8474d7794092c26da6dba14 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55b8bba1600b4e3891550da47d51f7f8 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/854137dcebb14526ae54f571fe07a6f9 : SUCCESS in 7m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/351c552cdb424cbf92155c6a4f35b551 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4ccfd12eb704a429f3101168abc3de8 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef19950f22584bcebcecad183f6c0d61 : SUCCESS in 7m 08s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/469e1bc5d7cf4a4aadc5b5e15833c6d2 : SUCCESS in 47m 51s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e5e4e4d33b0c4222949385b981f6ff99 : SUCCESS in 52m 20s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/a7c5b150dff749e7bc757bcbb5f96494 : SUCCESS in 1h 06m 04s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/b57569b7c12643eabf7bbf69a33e64ef : FAILURE in 39m 21s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/bfc9b00a31e0480bb604d2df7bf3b879 : SUCCESS in 42m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66eb787b11b6b011a53d43144929a748d03bb544","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-14 05:34:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41a98be0066f97fa08e22db2fef6d7c42dcedf56","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2020-03-20 07:35:57.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\ncapsule_uuid is the uuid of sandbox contaier which providers network.\nSo sandbox would be regarded as a Container create by zun.","accounts_in_message":[],"_revision_number":1},{"id":"01a520292ee492bff4c0afcb1a6c23bb8d6c799f","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2020-03-20 09:39:32.000000000","message":"Patch Set 1:\n\nIMO，move network_attach/detach/list to a new file.\nThen the operations would be used by Container and Capsule.","accounts_in_message":[],"_revision_number":1},{"id":"4570599daac8cfa73b772802dac2e650677b268d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-20 13:50:51.000000000","message":"Patch Set 1:\n\n\u003e IMO，move network_attach/detach/list to a new file.\n \u003e Then the operations would be used by Container and Capsule.\n\nOK. I will modify this patch accordingly.","accounts_in_message":[],"_revision_number":1},{"id":"38147912f9429c25e1c00224ea0540282fbc9669","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-29 17:03:52.000000000","message":"Patch Set 1:\n\nA BP for this feature: https://blueprints.launchpad.net/zun/+spec/attach-detach-network-for-capsule","accounts_in_message":[],"_revision_number":1},{"id":"c032dc16f72036440c930b862cf0284a1ecf12ec","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-30 02:06:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89eee421bf663e7bc01ac6792b4e28311615cd50","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-30 02:06:58.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"a8197c66962c62b5d4ad1e72cd2339b7c3ba6204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 03:07:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7059eff1b9e4df5af03c58c337dc757 : FAILURE in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11422820aac64cc0a06c29d50af1a059 : FAILURE in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e202994ff8f34ebbadb6124ec45ea6f4 : FAILURE in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd7164b2e15143de9d4fe1834667d79e : FAILURE in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f35f1c1372e94819a0ee633f332a415c : FAILURE in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1556dcc2de6468a80ced528fa5c55b8 : FAILURE in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fa351d86e0947518be1457035126094 : SUCCESS in 5m 46s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/54b9c51b616142d9b1e89075a012b163 : SUCCESS in 45m 25s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/6a9f506015734574a14bb1d699276a54 : SUCCESS in 46m 46s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/58efcbd788194b70829bb3fdb034ad71 : SUCCESS in 59m 11s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/4da197f80dcc4e5e847c45c0537875e5 : SUCCESS in 30m 31s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/36005e7517094ed88e998e7314a91e47 : SUCCESS in 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16789cc845f8928d302ee12197690e0d5f50f2c0","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-30 03:25:27.000000000","message":"Patch Set 2:\n\nI looks a bit more closer to this feature. As you know, we are migrating from docker to CRI runtime. Docker supports attach/detach network to a container. However, CRI runtime doesn\u0027t support attach/detach network to a pod.\n\nTherefore, it is quite hard to add this feature, because CRI doesn\u0027t support it.","accounts_in_message":[],"_revision_number":2},{"id":"ab61b5765ef06f10022ea4c9c5170a2a01f3e0ab","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-02 03:21:59.000000000","message":"Patch Set 2:\n\nI will revisit this patch","accounts_in_message":[],"_revision_number":2},{"id":"df6f87f5627b36bdbc66a409be6e9f4e5a1ede8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-12-30 12:29:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"00b0ac4565fa3aa9dd91f7a8ebbc7f060830c6a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-30 13:32:17.000000000","message":"Patch Set 3: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47be1274ecb7465c87970bbd8cb393c2 : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac665cdb535846fcbb3513933eab4f11 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bada65be1383408c9d5fd76947e5d2a0 : FAILURE in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f8babfc478c4124835538d99bb8f030 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e9a363636064d04acc59afb8912cc9e : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79701e6bf44841d998792bf5006ca8ca : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38960cc91a954716995585ee5599e05d : SUCCESS in 5m 37s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/1eef341624494cb1afe4a5e73dd6cec6 : SUCCESS in 37m 06s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/1501f7a572114100b94d2421fe1bdfaa : SUCCESS in 42m 15s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/249696ca815f4427899e4e7a5df7da3d : SUCCESS in 1h 01m 29s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/7d175a7af9c34991a2ee3a8a8b6bb5af : FAILURE in 43m 01s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/d3d49331a6174da1a51dfcc5eae80c91 : SUCCESS in 38m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"966a9226b5cfa7c829f572741ca8bab473de9852","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-12-31 10:55:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f43b84ce0dcaab204c3768fc046a8d813953279b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-31 11:34:20.000000000","message":"Patch Set 4: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/386cf8a9020f4cecbe515c4b6010cdb3 : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec25a359269f461ba6f32e2907a89eee : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6b553dc37b48308d2df4c181d7207a : FAILURE in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b848f184f29c42d78333847d83165e30 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87998f5aefed4403a5954694a37b8319 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61c7b980c4764e9a91c5bf1d6bb59bf1 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/296f7582c92a49d78de0fcedddbb47e3 : SUCCESS in 8m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/462823b86807474c873764bdf07c3bc2 : FAILURE in 24m 25s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e3b5d0881df442628aff8dee81470abd : FAILURE in 32m 26s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/a8f8367e86234b72a141afaa4883af33 : FAILURE in 29m 20s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/c5bea14b3da04f6b85f1bcfd69e1fdcd : FAILURE in 33m 34s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/3cc758f7af6e4f13aa3ba11fd8cb0e81 : FAILURE in 29m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"76ff90b1c95ce7c31234020d5e2d2037412f5db7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-01-01 07:39:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6871d7fa3bee2d13f97f8bb127e24e505aec44b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-01 08:25:39.000000000","message":"Patch Set 5: Verified-1\n\n(36 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/703a084d2a0b41a18b848d801d74e760 : SUCCESS in 7m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44c75863ed0d4346894c1b98ee78d490 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edd3c433512346c197ae4fca12a6e14f : FAILURE in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/feacfc51a1064cb380f817e58b993afa : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/678152d128104506a0feb621f75c2ae4 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cd620950ae44192af0d34f934035ca6 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a26eb803d954e4a903baf14f65d9735 : SUCCESS in 6m 26s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/724e7c2fa6d64d26a7cc2f2f4acc532b : FAILURE in 31m 03s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/d1181d5fdac248fa95a4d985bebfb982 : FAILURE in 32m 29s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/1d7654b2b4894df3bb52b64ba8236567 : FAILURE in 38m 29s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/7e949431059e427d91d3c84a6a62fba1 : FAILURE in 31m 57s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/9438af322fdd4249b412476276bf1738 : FAILURE in 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5054d633ac9fd39d30862470a791e72c1be01b6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-01-04 08:53:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"58b27a6c3c06479189b9ff4af7aacbf53ff3ff3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 09:31:09.000000000","message":"Patch Set 6: Verified-1\n\n(48 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd7a3301bf574018b2ac874c6fb2162c : FAILURE in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18ba92f54f504286b0e157eda98e429a : FAILURE in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/906ed7e8ad07411f8260dd74f6400d20 : FAILURE in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ed513f904484c9e8e773f5e120f6386 : FAILURE in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/927a46f2dad541dc96613df1c38ebb9c : FAILURE in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ad77c628f704cbd9c2eaed270c77d02 : FAILURE in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c1b597a0be4a8398554f1a7e85b6e4 : FAILURE in 5m 52s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b4ff9f1820d249708409ff9f8ea3d35b : FAILURE in 30m 59s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/9acf4839495c4fffbe37dc82687354f9 : FAILURE in 35m 34s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/9db29b6e2a1c46cba6d7a45c3d01036c : FAILURE in 34m 45s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/3e7633c3e8dc4a589607d88dc9f47436 : FAILURE in 26m 59s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/cbf412090d914e59bb13c8793b350259 : FAILURE in 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5393d1735664b8bae1bf3181d9f924d9f60f8527","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-03-21 05:02:58.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"957d967e5079217fc3df256c9cc06bb4f8ebeb3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-21 05:40:37.000000000","message":"Patch Set 7: Verified-1\n\n(48 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c0d43378c644b40aa3909796ca1ba99 : FAILURE in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1227b5a3e43142a18290cc6fe00aea5a : FAILURE in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4315521b899e43419f1350f9e783cd54 : FAILURE in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/421853df7ec440c5bf2ca91faf9376b8 : FAILURE in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0312bb439e664be7ad2a2de00876bc3d : FAILURE in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0bb701ce4454376b6c63eb621709f71 : FAILURE in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2045e56561f0469e9728407c7fae2e54 : FAILURE in 6m 25s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9bf4bd247b314a67ae9d1cf1250ed510 : FAILURE in 29m 59s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e34b465289f6495195df3d84b2eb75ad : FAILURE in 29m 00s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/15fe2c7fb530490c98730b0d3d1b8bc7 : FAILURE in 36m 10s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/05de563a088a40668319d611765a2e40 : FAILURE in 32m 02s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/6e15ff849f354194903a4cb5b2aeff96 : FAILURE in 26m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0b72b24c3c9da96fc3bf6dd510a34d3adabb9c19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-03-28 10:06:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"85990d65e7beb6b70d93be40fda60fec620934df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-28 10:56:14.000000000","message":"Patch Set 8: Verified-1\n\n(48 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f562f3e0bfe04d8eb4c8d9117ddc21b1 : FAILURE in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0c5edbc40ae41a1851968cd20b8a6a0 : FAILURE in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7afc34cfb8c64a77979cbcb46758eee4 : FAILURE in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aba656ab44449f0a0b7ece84a1cb8fb : FAILURE in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1474554c7e9b42a794ed01fb88a35dd0 : FAILURE in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3c6127bca744c4a92ad00507ef4877b : FAILURE in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/269ea1494ed74570987387587b0ad268 : FAILURE in 6m 24s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/cce3acbe170447b9a2dbde6b0984e9e9 : FAILURE in 34m 08s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/cad83706797642c7aa5191224f1e13f0 : FAILURE in 34m 22s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/09bb0cf246b24bacb491479ab2bb6242 : FAILURE in 43m 18s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/637610405a2d486493f417272e6e6338 : FAILURE in 36m 55s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/8d7ee88dd7f348cb82cb18fa53984396 : FAILURE in 31m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2f72d0fb19928d6e00ccdea90ba32549760629ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-03-28 12:36:39.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ac64240f487d54065d1a1d4e08ccd209cb2fc4f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-28 13:30:23.000000000","message":"Patch Set 9: Verified-1\n\n(41 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c69f4e9b0951439f96d1759f978e50f3 : FAILURE in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bea1b28ff974164a56cec1fc9224efa : FAILURE in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa420f3473649ccb8d78963b6240649 : FAILURE in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9b9437e675a464583d9f5a362d2eeb2 : FAILURE in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26f19c7b49af41709c544739903b02a3 : FAILURE in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7073c4525de440bc8f1563a4bd604a9a : FAILURE in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7b48753d0941078ae175d57e3c52f2 : FAILURE in 6m 27s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/858d17d0ade04e0da6e54ab71ae7ef9f : FAILURE in 35m 09s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/c0c3be2f18b149b0aa0dab8d5c15e875 : FAILURE in 34m 30s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b288413999034336878dbd0c9a5cabb9 : FAILURE in 47m 40s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/58e1ad24d8d14b34a9f744a381ed3983 : FAILURE in 37m 43s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/ea417f6a758b4a05a4a10729dc77cc85 : FAILURE in 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"93053a1b6f0a92aec5e77006323f98e30a6108f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-07-10 08:24:00.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9cc33aa328483b8995f2445605a4bd25d8917e8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 08:54:50.000000000","message":"Patch Set 10: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcaa75818f244f6bb8861acee400e4e6 : FAILURE in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/495b9d2e0cff47cb9642b40c956d8fe0 : FAILURE in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d902a5c2bd24527acdc9f4afe89e2fe : FAILURE in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22e9a78ce18e488fa53c195a0fc9c5e6 : FAILURE in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53cde38ab9e641bc8c62e44e489ce608 : FAILURE in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43445f75ace14b3a91504a7c252429f4 : FAILURE in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfe57a9eb61148a1b9dd535ed86ae7c5 : FAILURE in 7m 04s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/67059bd9777e4bf7a795717f2c48b603 : FAILURE in 24m 20s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/b5855773d9a74d418c18454f8379c6ee : FAILURE in 23m 47s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/58e82ff041b445ab9fffcc8acae6ac11 : FAILURE in 29m 46s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/ae6847398d754001a09a08724fe8f270 : FAILURE in 26m 01s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/35d4171238044c14a746509c039a75f6 : FAILURE in 22m 20s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"486b8c014ca56d4a0d9b347341995486c617cdde","revisions":{"aef594f105b2db18a33476e68b9891722bcd8ff1":{"kind":"REWORK","_number":1,"created":"2020-02-13 04:47:06.000000000","uploader":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"ref":"refs/changes/26/707526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/26/707526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/26/707526/1"}}},"commit":{"parents":[{"commit":"6fbf52ae2448bfa75635060d93cb5a94c0d2caab","subject":"Merge \"Introduce a CRI driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6fbf52ae2448bfa75635060d93cb5a94c0d2caab"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-02-13 04:48:07.000000000","tz":480},"committer":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-02-13 04:50:09.000000000","tz":480},"subject":"support to attach/detach network to capsule","message":"support to attach/detach network to capsule\n\nChange-Id: Ife71fb32e0a091aaef30e97e626be307ff851356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/aef594f105b2db18a33476e68b9891722bcd8ff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/aef594f105b2db18a33476e68b9891722bcd8ff1"}]},"branch":"refs/heads/master"},"e8e56dc1c9a15008ae60c69d3b809a1300d7d6ac":{"kind":"REWORK","_number":2,"created":"2020-03-30 02:06:44.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/26/707526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/26/707526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/26/707526/2"}}},"commit":{"parents":[{"commit":"33b96a81a75f0593c443daa1220b7cfc30b8540f","subject":"Fix autogenerated db migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/33b96a81a75f0593c443daa1220b7cfc30b8540f"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-02-13 04:48:07.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-30 02:06:31.000000000","tz":0},"subject":"[WIP] support to attach/detach network to capsule","message":"[WIP] support to attach/detach network to capsule\n\nChange-Id: Ife71fb32e0a091aaef30e97e626be307ff851356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e8e56dc1c9a15008ae60c69d3b809a1300d7d6ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e8e56dc1c9a15008ae60c69d3b809a1300d7d6ac"}]},"branch":"refs/heads/master"},"487203f3b5117bc12c4d7a560c11585c4389ebc3":{"kind":"REWORK","_number":3,"created":"2020-12-30 12:29:36.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/26/707526/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/26/707526/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/26/707526/3"}}},"commit":{"parents":[{"commit":"8e7e4eb98608ecaa362b4d442b6550585f69112f","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8e7e4eb98608ecaa362b4d442b6550585f69112f"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-02-13 04:48:07.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-30 12:29:33.000000000","tz":480},"subject":"[WIP] support to attach/detach network to capsule","message":"[WIP] support to attach/detach network to capsule\n\nChange-Id: Ife71fb32e0a091aaef30e97e626be307ff851356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/487203f3b5117bc12c4d7a560c11585c4389ebc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/487203f3b5117bc12c4d7a560c11585c4389ebc3"}]},"branch":"refs/heads/master"},"4bc83acfbe38013f7ac3c24f424164f4bcedbec8":{"kind":"REWORK","_number":4,"created":"2020-12-31 10:55:41.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/26/707526/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/26/707526/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/26/707526/4"}}},"commit":{"parents":[{"commit":"8e7e4eb98608ecaa362b4d442b6550585f69112f","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8e7e4eb98608ecaa362b4d442b6550585f69112f"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-02-13 04:48:07.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-31 10:55:30.000000000","tz":480},"subject":"[WIP] support to attach/detach network to capsule","message":"[WIP] support to attach/detach network to capsule\n\nChange-Id: Ife71fb32e0a091aaef30e97e626be307ff851356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4bc83acfbe38013f7ac3c24f424164f4bcedbec8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4bc83acfbe38013f7ac3c24f424164f4bcedbec8"}]},"branch":"refs/heads/master"},"3634701e5cad0d64df8a211a6b3f8f64c060eeeb":{"kind":"REWORK","_number":5,"created":"2021-01-01 07:39:59.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/26/707526/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/26/707526/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/26/707526/5"}}},"commit":{"parents":[{"commit":"8e7e4eb98608ecaa362b4d442b6550585f69112f","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8e7e4eb98608ecaa362b4d442b6550585f69112f"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-02-13 04:48:07.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-01-01 07:39:47.000000000","tz":480},"subject":"[WIP] support to attach/detach network to capsule","message":"[WIP] support to attach/detach network to capsule\n\nChange-Id: Ife71fb32e0a091aaef30e97e626be307ff851356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3634701e5cad0d64df8a211a6b3f8f64c060eeeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3634701e5cad0d64df8a211a6b3f8f64c060eeeb"}]},"branch":"refs/heads/master"},"15905311012cd63edda17fa817334d4ca352de78":{"kind":"REWORK","_number":6,"created":"2021-01-04 08:53:56.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/26/707526/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/26/707526/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/26/707526/6"}}},"commit":{"parents":[{"commit":"8e7e4eb98608ecaa362b4d442b6550585f69112f","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8e7e4eb98608ecaa362b4d442b6550585f69112f"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-02-13 04:48:07.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-01-04 08:53:44.000000000","tz":480},"subject":"[WIP] support to attach/detach network to capsule","message":"[WIP] support to attach/detach network to capsule\n\nChange-Id: Ife71fb32e0a091aaef30e97e626be307ff851356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/15905311012cd63edda17fa817334d4ca352de78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/15905311012cd63edda17fa817334d4ca352de78"}]},"branch":"refs/heads/master"},"1332232b034716f7a69b78366c81891392c94c3c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-03-21 05:02:58.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/26/707526/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/26/707526/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/26/707526/7"}}},"commit":{"parents":[{"commit":"6ac7a3ee5322164995248f4534f807e897991180","subject":"Remove __unicode__() from ZunException","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6ac7a3ee5322164995248f4534f807e897991180"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-02-13 04:48:07.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-03-21 05:02:55.000000000","tz":480},"subject":"[WIP] support to attach/detach network to capsule","message":"[WIP] support to attach/detach network to capsule\n\nChange-Id: Ife71fb32e0a091aaef30e97e626be307ff851356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1332232b034716f7a69b78366c81891392c94c3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1332232b034716f7a69b78366c81891392c94c3c"}]},"branch":"refs/heads/master"},"c7b5e35b4daef45566294125f60c35bb3268688a":{"kind":"REWORK","_number":8,"created":"2021-03-28 10:06:47.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/26/707526/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/26/707526/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/26/707526/8"}}},"commit":{"parents":[{"commit":"917868f5fe02ff419fd35c5d9332f45a064ed385","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/917868f5fe02ff419fd35c5d9332f45a064ed385"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-02-13 04:48:07.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-03-28 10:06:44.000000000","tz":480},"subject":"[WIP] support to attach/detach network to capsule","message":"[WIP] support to attach/detach network to capsule\n\nChange-Id: Ife71fb32e0a091aaef30e97e626be307ff851356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c7b5e35b4daef45566294125f60c35bb3268688a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c7b5e35b4daef45566294125f60c35bb3268688a"}]},"branch":"refs/heads/master"},"536fecb49e46d499986454cd2d98e045697764d2":{"kind":"REWORK","_number":9,"created":"2021-03-28 12:36:39.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/26/707526/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/26/707526/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/26/707526/9"}}},"commit":{"parents":[{"commit":"917868f5fe02ff419fd35c5d9332f45a064ed385","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/917868f5fe02ff419fd35c5d9332f45a064ed385"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-02-13 04:48:07.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-03-28 12:36:25.000000000","tz":480},"subject":"[WIP] support to attach/detach network to capsule","message":"[WIP] support to attach/detach network to capsule\n\nChange-Id: Ife71fb32e0a091aaef30e97e626be307ff851356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/536fecb49e46d499986454cd2d98e045697764d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/536fecb49e46d499986454cd2d98e045697764d2"}]},"branch":"refs/heads/master"},"486b8c014ca56d4a0d9b347341995486c617cdde":{"kind":"REWORK","_number":10,"created":"2021-07-10 08:24:00.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/26/707526/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/26/707526/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/26/707526/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/26/707526/10"}}},"commit":{"parents":[{"commit":"abd1e5c9f34c3b3c141600f0386e77ab41f7e89a","subject":"Respect allocation ratio on container update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/abd1e5c9f34c3b3c141600f0386e77ab41f7e89a"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-02-13 04:48:07.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-07-10 08:23:57.000000000","tz":480},"subject":"[WIP] support to attach/detach network to capsule","message":"[WIP] support to attach/detach network to capsule\n\nChange-Id: Ife71fb32e0a091aaef30e97e626be307ff851356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/486b8c014ca56d4a0d9b347341995486c617cdde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/486b8c014ca56d4a0d9b347341995486c617cdde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
