)]}'
{"id":"openstack%2Fzun~707948","triplet_id":"openstack%2Fzun~master~I61878e646add132e0b685ef3b4f4393a57089bb0","project":"openstack/zun","branch":"master","topic":"bp/add-support-cri-runtime","hashtags":[],"change_id":"I61878e646add132e0b685ef3b4f4393a57089bb0","subject":"Add CRI to installation guide","status":"MERGED","created":"2020-02-15 19:49:37.000000000","updated":"2020-02-22 02:28:12.000000000","submitted":"2020-02-22 02:26:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707948-1582338408821-7c03db43","meta_rev_id":"b3a7de4e33ac4a94a69c7dfdddab41a8c5f36c16","_number":707948,"virtual_id_number":707948,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-22 02:26:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2020-02-22 02:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-17 11:29:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-02-22 02:15:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 11:29:00.000000000","updated_by":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"reviewer":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"state":"REVIEWER"},{"updated":"2020-02-22 02:15:46.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2020-02-22 02:26:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"069594177005a3a7d9ca492c67e88e911f42a9c8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-15 19:49:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64ff800e0acf31a5d2ec7d6db68e05147a665f3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 20:00:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc32adbd78ec478cb5231e0208bf2c5b : FAILURE in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4949f1504df541c8a00e4ba17283dc06 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"52518ecc75d7909a808df70db7f94e51d00eaec8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-16 19:59:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8391c6a027776e89f58d8adecbdbb63a54547f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 20:11:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c59eaa59db1457c986cd5c59b052757 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e79f232091e44ae9b7aabf8a9838b69 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":2},{"id":"593694d1b77683f620b2c4742a8a7adffef51b8a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 00:58:10.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"5bd430c1ebb6dd3f793b02eefd1b5bdc36bad659","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 02:49:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"65b76e7ee1401ab147b02ba51f9991a7c13914ed","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 02:51:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4d8f1d43598668e5ecab0e162d7e585371b7e691","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 02:52:01.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"c30d48ccd815e7c4d99e9c4869f6f578071d6c1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 03:00:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f77d8e733e4e43f5830daadfab6064d3 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a11f680f7244802b0e102bafd42db58 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":5},{"id":"ef6599a0257c3d2edef9fce271f2c77eade460bb","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 05:04:44.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"61838a66cd19f0839871e81340b5861c913b7b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 05:11:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8ab154841b44953816b781379197dab : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0e29d50637c4c13b4fbced58f3f424c : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":6},{"id":"0e37624843559e82aab5aa68eeba6f30962abeb9","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2020-02-17 11:29:00.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"69ff389e47d1fe5a2c57f6a6ede51f627f95e5fb","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2020-02-22 02:15:46.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"20416b569536ff4a0c71792b1c2b8b384816cbbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 02:15:57.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1c9f9f060d457c9f872826220e3538a16d49c400","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 02:26:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"e8353d9d3dc287472af6a343c0d58b39b12d56a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 02:26:48.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8adec10e8c64a8790caa365f89f1901 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5932de441744b938b73911086b02e60 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":6},{"id":"58b4c7b278191e65899598a49eafde72e2987f22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 02:28:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da93ce3cef004cf78c588db05b4c0782 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"eb1c32951b0d48e5956eb80415944b5ac28fd61a","revisions":{"545d8b6bc34cd524e0786787d7019f015f97dc83":{"kind":"REWORK","_number":1,"created":"2020-02-15 19:49:37.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/48/707948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/48/707948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/48/707948/1"}}},"commit":{"parents":[{"commit":"6fbf52ae2448bfa75635060d93cb5a94c0d2caab","subject":"Merge \"Introduce a CRI driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6fbf52ae2448bfa75635060d93cb5a94c0d2caab"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-15 17:42:21.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-15 19:49:28.000000000","tz":0},"subject":"[WIP] Add CRI to installation guide","message":"[WIP] Add CRI to installation guide\n\nChange-Id: I61878e646add132e0b685ef3b4f4393a57089bb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/545d8b6bc34cd524e0786787d7019f015f97dc83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/545d8b6bc34cd524e0786787d7019f015f97dc83"}]},"branch":"refs/heads/master"},"03f2cc96c0e87538fc9d45eca39f784afeeb5de2":{"kind":"REWORK","_number":2,"created":"2020-02-16 19:59:47.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/48/707948/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/48/707948/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/48/707948/2"}}},"commit":{"parents":[{"commit":"6fbf52ae2448bfa75635060d93cb5a94c0d2caab","subject":"Merge \"Introduce a CRI driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6fbf52ae2448bfa75635060d93cb5a94c0d2caab"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-15 17:42:21.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-16 19:59:40.000000000","tz":0},"subject":"[WIP] Add CRI to installation guide","message":"[WIP] Add CRI to installation guide\n\nChange-Id: I61878e646add132e0b685ef3b4f4393a57089bb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/03f2cc96c0e87538fc9d45eca39f784afeeb5de2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/03f2cc96c0e87538fc9d45eca39f784afeeb5de2"}]},"branch":"refs/heads/master"},"3287497480b9ebb062874c99fcda23e61715745d":{"kind":"REWORK","_number":3,"created":"2020-02-17 02:49:19.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/48/707948/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/48/707948/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/48/707948/3"}}},"commit":{"parents":[{"commit":"6fbf52ae2448bfa75635060d93cb5a94c0d2caab","subject":"Merge \"Introduce a CRI driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6fbf52ae2448bfa75635060d93cb5a94c0d2caab"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-15 17:42:21.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 02:49:07.000000000","tz":0},"subject":"Add CRI to installation guide","message":"Add CRI to installation guide\n\nChange-Id: I61878e646add132e0b685ef3b4f4393a57089bb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3287497480b9ebb062874c99fcda23e61715745d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3287497480b9ebb062874c99fcda23e61715745d"}]},"branch":"refs/heads/master"},"499ffbe0920531974c342f24980c9726beaea3f3":{"kind":"REWORK","_number":4,"created":"2020-02-17 02:51:24.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/48/707948/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/48/707948/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/48/707948/4"}}},"commit":{"parents":[{"commit":"6fbf52ae2448bfa75635060d93cb5a94c0d2caab","subject":"Merge \"Introduce a CRI driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6fbf52ae2448bfa75635060d93cb5a94c0d2caab"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-15 17:42:21.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 02:51:17.000000000","tz":0},"subject":"Add CRI to installation guide","message":"Add CRI to installation guide\n\nChange-Id: I61878e646add132e0b685ef3b4f4393a57089bb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/499ffbe0920531974c342f24980c9726beaea3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/499ffbe0920531974c342f24980c9726beaea3f3"}]},"branch":"refs/heads/master"},"932128e225ec4c171ce7fd23169d8ccd12b9f3e1":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-02-17 02:52:01.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/48/707948/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/48/707948/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/48/707948/5"}}},"commit":{"parents":[{"commit":"6fbf52ae2448bfa75635060d93cb5a94c0d2caab","subject":"Merge \"Introduce a CRI driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6fbf52ae2448bfa75635060d93cb5a94c0d2caab"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-15 17:42:21.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 02:51:41.000000000","tz":0},"subject":"Add CRI to installation guide","message":"Add CRI to installation guide\n\nImplements: blueprint add-support-cri-runtime\nChange-Id: I61878e646add132e0b685ef3b4f4393a57089bb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/932128e225ec4c171ce7fd23169d8ccd12b9f3e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/932128e225ec4c171ce7fd23169d8ccd12b9f3e1"}]},"branch":"refs/heads/master"},"eb1c32951b0d48e5956eb80415944b5ac28fd61a":{"kind":"NO_CHANGE","_number":6,"created":"2020-02-17 05:04:44.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/48/707948/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/48/707948/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/48/707948/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/48/707948/6"}}},"commit":{"parents":[{"commit":"6fbf52ae2448bfa75635060d93cb5a94c0d2caab","subject":"Merge \"Introduce a CRI driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6fbf52ae2448bfa75635060d93cb5a94c0d2caab"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-15 17:42:21.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 05:04:38.000000000","tz":0},"subject":"Add CRI to installation guide","message":"Add CRI to installation guide\n\nImplements: blueprint add-support-cri-runtime\nChange-Id: I61878e646add132e0b685ef3b4f4393a57089bb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/eb1c32951b0d48e5956eb80415944b5ac28fd61a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/eb1c32951b0d48e5956eb80415944b5ac28fd61a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
