)]}'
{"id":"openstack%2Fzun~708052","triplet_id":"openstack%2Fzun~master~Ic64853d9a5ec3a5b30af372dc551bad844541aca","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ic64853d9a5ec3a5b30af372dc551bad844541aca","subject":"Update installation guide for using python3","status":"MERGED","created":"2020-02-17 04:43:17.000000000","updated":"2020-03-07 17:24:15.000000000","submitted":"2020-03-07 17:22:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708052-1583601771356-74ea99eb","meta_rev_id":"11614aefaee92501c4fbbd06bfeadf141c3c17a8","_number":708052,"virtual_id_number":708052,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-07 17:22:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-03-07 17:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-03-07 17:12:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-07 17:22:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d704cef5b17e9e7446b048c0b5f4536c3a4ab249","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 04:43:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c658631f33767790d2cd0ef3862127df80814666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 04:52:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/955550d03b8b43e597a4b3562deda837 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c92d5a34e7f740a995c3d34f755b753e : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ac57f9332913b8610f57ceeccb372fd0e4283e50","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 05:04:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0b50dd4e9f3929b6ee6353e93cd08c8c57f14d4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 05:15:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/763d7d0b1a1240da9c5c5347b40543fa : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57406382683743e7bcf0e70b5c0bdc39 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":2},{"id":"95b84df58f1723a72b856b869e7a27ce7551452b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-07 17:12:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1838c647a1994d38626fdd8df3d2160aa0946796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 17:12:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e9bddfc9b5f62add21a2b052f8637efa24263d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 17:22:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/118993ea0d0443888d59ba3bf61c6bb2 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e54c3746e545468d0a3b28e7460909 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d456623c0187b616f824a92dd5d37e16ae7cfb6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 17:22:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"11614aefaee92501c4fbbd06bfeadf141c3c17a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 17:24:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c5b509db0f4094b80c583e1f1cef1f : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"edbd6cfd9965da1925e3275b9d76788d968c1cec","revisions":{"1a5744c8629c43a98a8dfdcb35fd008af8c5205a":{"kind":"REWORK","_number":1,"created":"2020-02-17 04:43:17.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/52/708052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/52/708052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/52/708052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/52/708052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/52/708052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/52/708052/1"}}},"commit":{"parents":[{"commit":"932128e225ec4c171ce7fd23169d8ccd12b9f3e1","subject":"Add CRI to installation guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/932128e225ec4c171ce7fd23169d8ccd12b9f3e1"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 04:42:48.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 04:42:48.000000000","tz":0},"subject":"Update installation guide for using python3","message":"Update installation guide for using python3\n\nChange-Id: Ic64853d9a5ec3a5b30af372dc551bad844541aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1a5744c8629c43a98a8dfdcb35fd008af8c5205a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1a5744c8629c43a98a8dfdcb35fd008af8c5205a"}]},"branch":"refs/heads/master"},"edbd6cfd9965da1925e3275b9d76788d968c1cec":{"kind":"NO_CHANGE","_number":2,"created":"2020-02-17 05:04:44.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/52/708052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/52/708052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/52/708052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/52/708052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/52/708052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/52/708052/2"}}},"commit":{"parents":[{"commit":"eb1c32951b0d48e5956eb80415944b5ac28fd61a","subject":"Add CRI to installation guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/eb1c32951b0d48e5956eb80415944b5ac28fd61a"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 04:42:48.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 05:04:38.000000000","tz":0},"subject":"Update installation guide for using python3","message":"Update installation guide for using python3\n\nChange-Id: Ic64853d9a5ec3a5b30af372dc551bad844541aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/edbd6cfd9965da1925e3275b9d76788d968c1cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/edbd6cfd9965da1925e3275b9d76788d968c1cec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
