)]}'
{"id":"openstack%2Fzun~713871","triplet_id":"openstack%2Fzun~master~Ib67746455a33306bb6d09a7fbfc8f543a0938c52","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ib67746455a33306bb6d09a7fbfc8f543a0938c52","subject":"failed to delele capsule when sandbox is removed by docker","status":"MERGED","created":"2020-03-19 12:59:38.000000000","updated":"2020-03-19 15:41:28.000000000","submitted":"2020-03-19 15:39:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713871-1584632360743-46f3065f","meta_rev_id":"9f4b77f282bbdb2d7376a7683fc3e87d989c728b","_number":713871,"virtual_id_number":713871,"owner":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 15:39:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-03-19 14:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-03-19 14:22:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 14:22:01.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-03-19 15:39:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9750fdb728c622582db4d893fcfd19748b3247c4","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2020-03-19 12:59:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb66413481f3a48e6307bce08a2bd7b0b8aa28b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:09:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/712220bc475945438fb269f6b58a4df1 : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25904ac2c4e34ce9b1ecf6f0a1298187 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cfabcf1cb324ccf9285da10a2e84bbf : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd93d5b2942f402c97662cbc77365626 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7de263d36884496e99040b3929994f97 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c55f6cf5f404de382e8d4dba11dc947 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6de0e1af6d046798966a1070470324e : SUCCESS in 6m 54s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ecc3ec368b194890b00aded3e2dbc19b : SUCCESS in 58m 32s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/75db4d95fcff4eb69cae2ba49d494b10 : SUCCESS in 52m 58s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/36efe413c52948d3878de9a22a1d7357 : SUCCESS in 1h 03m 36s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/ce7e5316615c48b7879e18e01cc878f3 : SUCCESS in 37m 54s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/a5959bc733234f1a875281d6ccc10877 : SUCCESS in 44m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5b7842557175ba9780ede8d8bf795b3186cb93a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-19 14:22:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53db939fc5cd432ef371291f2c4e640ec4cc7814","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:22:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e64b7df4fdbae16e36ba0133196568bef0b4d43c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 15:39:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2fc3181c69b45da911f02e5630600ad : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5717a9de1cc43659f08827d8f8c2d61 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90bf14e0cf7f4201aaf33d3cbd91fa3a : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f178e2de0def4628b9f6e7f42121033a : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d294e604271f4cc1a34ca34a101ccaca : SUCCESS in 5m 35s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d88fe17e147c4951a10c4a406dfe8260 : SUCCESS in 50m 17s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/5b5355b4a8114f64a051c3cfbc4d158f : SUCCESS in 47m 45s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d7285b813efd4533bc10776e4cf8c868 : SUCCESS in 1h 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"575e3e790433c98757f1414278a90b6034fdcb12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 15:39:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9f4b77f282bbdb2d7376a7683fc3e87d989c728b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 15:41:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99e382666cb406a8ffb4e648769e0c4 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2f5fc4f5ba8ba7a5f5fa39c60b812ce8360771b","revisions":{"b2f5fc4f5ba8ba7a5f5fa39c60b812ce8360771b":{"kind":"REWORK","_number":1,"created":"2020-03-19 12:59:38.000000000","uploader":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"ref":"refs/changes/71/713871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/71/713871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/713871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/713871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/713871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/71/713871/1"}}},"commit":{"parents":[{"commit":"c2d56024d88725df5d5a94d8f011f2ef8bb88fba","subject":"delete inincontainer in capsule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c2d56024d88725df5d5a94d8f011f2ef8bb88fba"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-03-19 12:54:08.000000000","tz":480},"committer":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-03-19 12:54:08.000000000","tz":480},"subject":"failed to delele capsule when sandbox is removed by docker","message":"failed to delele capsule when sandbox is removed by docker\n\nChange-Id: Ib67746455a33306bb6d09a7fbfc8f543a0938c52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b2f5fc4f5ba8ba7a5f5fa39c60b812ce8360771b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b2f5fc4f5ba8ba7a5f5fa39c60b812ce8360771b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
