)]}'
{"id":"openstack%2Fzun~714042","triplet_id":"openstack%2Fzun~master~I2ce108f6bfc91e2aff9857593243a4c0412660bc","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I2ce108f6bfc91e2aff9857593243a4c0412660bc","subject":"Failed to format message","status":"MERGED","created":"2020-03-20 07:28:14.000000000","updated":"2020-03-20 15:17:51.000000000","submitted":"2020-03-20 15:01:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714042-1584716518744-5004bf0a","meta_rev_id":"74806fda80d39f4d80ebdcedd51dfcae2a7583b2","_number":714042,"virtual_id_number":714042,"owner":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 15:01:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-03-20 13:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-03-20 13:47:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 13:47:19.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-03-20 15:01:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c9791edc1303be134d75a6d6456795975375c4f","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2020-03-20 07:28:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed4d7482f12c00ec0e8a63ca3e7df62a560bb716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 09:00:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3da82c987ec34b01ad647072c9f8e23c : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb400b0de4c7455ab569d32a05a9a9c9 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42b5d75a01bb4a82b53708ffaf11cc14 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9a92571f6334ec9845e26d21cc8d6ca : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f23376bd65a44c1fbcb612584f9a6636 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c25bc0b2db914aa5bb0f790c627a7103 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe066341b9d455499eaa3ffc6228732 : SUCCESS in 5m 57s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/e88c2d6c11394c2080dd1f67bfbe69ca : SUCCESS in 45m 14s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/ece48664c2244f5a828f086fa7cfca3e : SUCCESS in 1h 08m 17s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f1b1f612da03432f88c16c28c2bf52d6 : SUCCESS in 1h 25m 42s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/fde62398f4d54780b1e042ba9bd3608e : SUCCESS in 54m 02s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/e60231430fee4a50be5092850e0a08f4 : SUCCESS in 59m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0be92378ed450a4eaa2390d76cbd5c71c62f8a0a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-20 13:47:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e149dcb69cc1498db332b7e5260918cc746efbf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 13:53:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b424131f23abaeb1b3892d1d872176c8cd4408bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 15:01:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a54b5e28ef7b46feaca27f6a8a7d2022 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32d91668cb124c61a5fea516a8cd7ea3 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9efc476ce224e719ab2023f0f988d73 : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/408f1fa33c3e4d53bcfc7a25f28d0893 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05c91aa2a124ab28f2d2fa468d337a0 : SUCCESS in 8m 29s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/dd4c8f11d776485193e48f49ee9af2c6 : SUCCESS in 49m 18s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/861338f249174aef9adc080d2d742c51 : SUCCESS in 47m 13s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/a4cf3a3f455640b4bef0f24a087c3bf1 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1f0d04da6afdf767f543f7dfd532d1f17adb0720","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 15:01:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"74806fda80d39f4d80ebdcedd51dfcae2a7583b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 15:17:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a1a19fa3e2944b5a1636c56d750c540 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5dc4ae5a0d8f9cd8509f17c8dab9d738d47bc7dd","revisions":{"5dc4ae5a0d8f9cd8509f17c8dab9d738d47bc7dd":{"kind":"REWORK","_number":1,"created":"2020-03-20 07:28:14.000000000","uploader":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"ref":"refs/changes/42/714042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/42/714042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/714042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/714042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/714042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/42/714042/1"}}},"commit":{"parents":[{"commit":"b2f5fc4f5ba8ba7a5f5fa39c60b812ce8360771b","subject":"failed to delele capsule when sandbox is removed by docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b2f5fc4f5ba8ba7a5f5fa39c60b812ce8360771b"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-03-20 07:15:59.000000000","tz":480},"committer":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-03-20 07:15:59.000000000","tz":480},"subject":"Failed to format message","message":"Failed to format message\n\nFile \"/usr/lib/python2.7/site-packages/oslo_privsep/daemon.py\", line 204, in remote_call\n  raise exc_type(*result[2])\nFile \"/usr/lib/python2.7/site-packages/zun/common/exception.py\", line 204, in __init__\n  self.message \u003d str(self.message) % kwargs\nValueError: unsupported format character \u0027)\u0027 (0x29) at index 345\n\nChange-Id: I2ce108f6bfc91e2aff9857593243a4c0412660bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5dc4ae5a0d8f9cd8509f17c8dab9d738d47bc7dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5dc4ae5a0d8f9cd8509f17c8dab9d738d47bc7dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
