)]}'
{"id":"openstack%2Fzun~714335","triplet_id":"openstack%2Fzun~master~I43e8b23355b52d2b55d36ba43701118efcf0e3ca","project":"openstack/zun","branch":"master","topic":"bp/support-cni","hashtags":[],"change_id":"I43e8b23355b52d2b55d36ba43701118efcf0e3ca","subject":"Add release note for Zun CNI plugin","status":"MERGED","created":"2020-03-22 21:38:36.000000000","updated":"2020-04-13 09:03:07.000000000","submitted":"2020-04-13 09:01:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714335-1586768470521-332a2ac9","meta_rev_id":"e3762a3c451b0ff233abbdf6af320aa15240d511","_number":714335,"virtual_id_number":714335,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-12 02:29:49.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 09:01:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2020-04-13 08:02:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-04-13 08:02:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 08:02:04.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2020-04-13 09:01:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b4193fadead69b2c11d6d6638ec9418bc3b9e0c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-22 21:38:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d20bc38f54bc79e3110377357534a9eccfa5466b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 22:46:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6164234c429a47fdb5d6cf8e64197ca1 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b58564d27e4432fac97ff9f3eef8b20 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fad4aff7be54e4dbc566139feefcab4 : SUCCESS in 4m 18s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/f687f0207c634a36b1fb72f29a5a5b00 : SUCCESS in 48m 52s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/1b5ec604610f4ace8fdf0b55d5821151 : SUCCESS in 51m 17s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/39fb99b44775484d9c78c895969188e8 : SUCCESS in 1h 05m 08s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/e3846b0a8611457e8460fa495bd3f43c : SUCCESS in 35m 38s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/73a783299f9c4e6484003b21595f1e14 : SUCCESS in 44m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"555a1774f79a50133eb21c3910fd018da4f05939","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-04-12 02:29:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b82569544bd5bd4c663144e361dcb33213877bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 03:34:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4dbb77336e48b89039f49c2256f93f : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b36dc3927ea41adb5569ef9b6f0c633 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95848aa3e80e4c11964a7d23fe20778f : SUCCESS in 3m 29s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a1d35cc419d14be0ba0743647824681e : SUCCESS in 48m 15s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/8c6974a69a26449c88163f2efc4a9aa5 : SUCCESS in 48m 25s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b68473429da14a999bc9e324beecf269 : SUCCESS in 59m 22s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/d6df67afbf7b451baac1f902c541dcf1 : SUCCESS in 33m 35s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/51db1dc891d74f8ea0b63be1a07a9ee6 : SUCCESS in 43m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b67d07ee72d782043e7e405ddac520dbec99917a","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2020-04-13 08:02:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"666278beae3b725183ef7928faad8c34579b064c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 08:02:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f2f4b2a5189f98700c1590515bdfa6cbeda4daf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:01:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f438a11c10004224ac71cf16375f8ec4 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e43a7767ba441f690325643dd2b123b : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd8ff5ae913f4216a5304d8f5dc7baae : SUCCESS in 3m 52s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/100a48eac9ea4aff852ea10ee7baefb2 : SUCCESS in 46m 54s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e8b3ea13ae7a43fd981dbe65b066b96a : SUCCESS in 44m 53s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/5f783dd390584f2382bb194194c12d30 : SUCCESS in 57m 28s","accounts_in_message":[],"_revision_number":1},{"id":"59d2b1e8e6e975b2a4eae0a82140dc7135cbc246","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:01:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8378dfc6c30d35374a21e1359fbe685c08448881","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:03:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eab03a1b6354ab5a20686d5af90376b : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/245c996ee77440b494de21863a5a52db : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"473ad2d69ea4f36ae8061a60e261c1ef0e16d619","revisions":{"473ad2d69ea4f36ae8061a60e261c1ef0e16d619":{"kind":"REWORK","_number":1,"created":"2020-03-22 21:38:36.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/35/714335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/35/714335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/35/714335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/35/714335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/35/714335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/35/714335/1"}}},"commit":{"parents":[{"commit":"c377cf8c17e266fff6ee98b418a105d5cafd19f4","subject":"Merge \"Enable CRI runtime in devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c377cf8c17e266fff6ee98b418a105d5cafd19f4"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-22 21:37:14.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-22 21:37:14.000000000","tz":0},"subject":"Add release note for Zun CNI plugin","message":"Add release note for Zun CNI plugin\n\nChange-Id: I43e8b23355b52d2b55d36ba43701118efcf0e3ca\nImplements: blueprint support-cni\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/473ad2d69ea4f36ae8061a60e261c1ef0e16d619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/473ad2d69ea4f36ae8061a60e261c1ef0e16d619"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
