)]}'
{"id":"openstack%2Fzun~714337","triplet_id":"openstack%2Fzun~master~I517a0ae4e03a6c8feed200dcda8a1370f42b8da4","project":"openstack/zun","branch":"master","topic":"bp/support-cni","hashtags":[],"change_id":"I517a0ae4e03a6c8feed200dcda8a1370f42b8da4","subject":"Add release note for CRI runtime support","status":"MERGED","created":"2020-03-22 23:09:19.000000000","updated":"2020-04-13 09:07:55.000000000","submitted":"2020-04-13 09:06:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714337-1586768785151-61dee02a","meta_rev_id":"4bbc4d7912f5c95b4429aecb0afc276b665e2010","_number":714337,"virtual_id_number":714337,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-12 02:29:39.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 09:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2020-04-13 08:04:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-04-13 08:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 08:04:40.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2020-04-13 09:06:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8509b6fb2eae12019700c38bb42b2fd95379493","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-22 23:09:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bc5b79997ff314bd4d2f3a360482fbc91c85b03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 00:13:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e536264528364020b903fc89403fdc35 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a4f795f543454bb631f92bd54fb450 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d4d9017364b463db2260f62e3ad3076 : SUCCESS in 4m 49s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/da5d2ef5cc75446ead99a36c63d40552 : SUCCESS in 47m 42s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/fbccaa1babb04e44b4fefb12fba6b845 : SUCCESS in 46m 49s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/1b255cbef44344dbb61414f669528a60 : SUCCESS in 1h 03m 04s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/cb446059201345d599d8f51a22d3dce5 : SUCCESS in 33m 26s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/04efbb4c3d834c7b8b5601c045b9496a : SUCCESS in 41m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24784cf198761c85c76d300dbb0ea10bfee91ad2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-04-12 02:29:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"411d245eecdd83df04288b33f9095f3b7c921bc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 03:35:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d59fe9f37a44cdf9b019be35d41487c : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be3ff1d5eced4060b4c98f147a4610f2 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce158d850999481f8583dbddbab8ea73 : SUCCESS in 3m 35s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/4d102f60c47245c0ab450335eadee1ce : SUCCESS in 45m 53s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/c7927c008a444b17b877f9cc33de200b : SUCCESS in 47m 41s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/7b407cb1c2f3450eadbf4f13e07f0be1 : SUCCESS in 1h 03m 59s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/a7ea88602df340228db7b4de2f2a2e9e : SUCCESS in 30m 28s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/9dbb2a2361914859ad6cff414a6fdb1c : SUCCESS in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8b4da2cf4da95610ad9aa451898ab6b90bf9c4a","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2020-04-13 08:04:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"38a0fc626fb39be5a3c750af35952384d4cf8006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 08:04:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0edb8b3c084bfc59ad4154917ecfaa3ec59868c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:06:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83b42ba388b24888b82ba8575345fa46 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65b8a12e5b346ab9a678667318a9327 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0279bf8043e041de87ea1640b668f78c : SUCCESS in 3m 35s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a3fad831ce184d938e815d190f10f2ef : SUCCESS in 44m 46s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/3619e509080c4f15ba779b33e957b14a : SUCCESS in 38m 57s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/5d3ac849e1264f968fc183185bbba960 : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f0ad0bad440b26dd793b04b1d63bccc71dac8222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:06:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9bab9ca5fdb9328b1aa097b60282641ba325a381","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:07:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a358b6c20c04d339d0dc50b4715056f : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/846e5a991c8f407088dca62f0b149b8b : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df62b3e1c18ccf05224643b9f1fd4ade029637ca","revisions":{"df62b3e1c18ccf05224643b9f1fd4ade029637ca":{"kind":"REWORK","_number":1,"created":"2020-03-22 23:09:19.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/37/714337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/37/714337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/714337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/714337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/714337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/37/714337/1"}}},"commit":{"parents":[{"commit":"c377cf8c17e266fff6ee98b418a105d5cafd19f4","subject":"Merge \"Enable CRI runtime in devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c377cf8c17e266fff6ee98b418a105d5cafd19f4"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-22 23:04:53.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-22 23:09:12.000000000","tz":0},"subject":"Add release note for CRI runtime support","message":"Add release note for CRI runtime support\n\nChange-Id: I517a0ae4e03a6c8feed200dcda8a1370f42b8da4\nImplements: blueprint support-cni\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/df62b3e1c18ccf05224643b9f1fd4ade029637ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/df62b3e1c18ccf05224643b9f1fd4ade029637ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
