)]}'
{"id":"openstack%2Fzun~716849","triplet_id":"openstack%2Fzun~master~I1c5f5a85af02d9f11e475b6088fe1d9d9d252083","project":"openstack/zun","branch":"master","topic":"update-hacking","hashtags":[],"change_id":"I1c5f5a85af02d9f11e475b6088fe1d9d9d252083","subject":"Update hacking for Python3","status":"MERGED","created":"2020-04-02 06:11:07.000000000","updated":"2020-04-02 14:19:31.000000000","submitted":"2020-04-02 14:16:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716849-1585836984203-9820b1aa","meta_rev_id":"aa2c56b851dabdbcc1bc328c1ad50893077e66c4","_number":716849,"virtual_id_number":716849,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 14:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-04-02 13:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-04-02 13:06:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 13:06:44.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-04-02 14:16:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb46116205892a98bd8ec06d63b33ec1cdeb411f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-02 06:11:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abdbf873a9577a3422ae1347083622f643f2cf00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 08:09:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2b58c45cc48549529130a34ac7a9d7da : SUCCESS in 8m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0592dac3f11c4f88835d94405aaa9da7 : SUCCESS in 5m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84ae84acbeeb4ce48b1c23a76309d955 : FAILURE in 7m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40e6c6002a7749179b7d84cc74336a42 : FAILURE in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23286cf5ecc54fccb8d87172ecc54768 : FAILURE in 15m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae93215d5f444f59832efbd9b31ac4fe : FAILURE in 9m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/257dd02ddfbd4fa382618c0630a374fe : FAILURE in 10m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e41bf479682948bfa34b66efdec1cf50 : FAILURE in 8m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb3e0375368c4d0ba72aacdeda76a2eb : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/666e7ef2f17042c6921ed59814618cee : SUCCESS in 6m 49s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/c18240561ad04e8280aeaabcd5d42639 : SUCCESS in 57m 07s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/fb09dc5c87d94dd4b244db8a757e647a : SUCCESS in 49m 56s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ebadd5e96ef74e12884cb5f394a50b43 : SUCCESS in 1h 01m 24s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/37b1838e348c4cb2bb3c8b708fe4c234 : SUCCESS in 38m 39s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/a84a00e1d044407f8d8474685b020e2a : SUCCESS in 42m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"463c3ef8a348a39b7f3055d4871f1970a9a592ee","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-02 08:22:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8694c4049adcd6011d6f5c8cb42a79262ac48a9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-02 08:28:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3f5cb2de07a6a5e7435fc51ec98d6e7e05980256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 09:29:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e175053f63474e358b7bc6085e9073c7 : SUCCESS in 5m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5481beceedd14530b564bf320492b2d4 : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23f1b7ff67e44697bdd9199b4461db8e : SUCCESS in 6m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6055a152c1d400396d2d8629f7c0d8a : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f538d33ee3bb4469b3632ef11cb9f084 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d31aec0361d4d109b83e03e777b75df : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cdeb1a5aa5264f928b7e25d680ce64a5 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30127a9f23874cb5a7de8fe8d5cacd92 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58acfbb0f70c495c8e8957385f98b037 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1f11eb443cf4602b5b621d3a07e59ea : SUCCESS in 4m 27s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/66e2375f9c9d4576aaaca02e4553c200 : SUCCESS in 48m 51s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e233d72079a04f1194a7eabfe1290de8 : SUCCESS in 47m 48s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2183d012621e4a37bee753b9c34a1d3e : SUCCESS in 58m 45s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/2ac7021700274164b16740e2ec1fcdca : SUCCESS in 32m 51s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/253c683e939c4743a6ef4474870ce927 : SUCCESS in 54m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f78f4bcb8ce48c5fd1d1cc5f9a7cb8dc4e593114","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-04-02 13:06:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks for the patch!","accounts_in_message":[],"_revision_number":3},{"id":"42630d7ec455b1f037d3a385749d3c82148ff5f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 13:06:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8902ec8924cb60ccf56828ca87d9bf5b606afcf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 14:16:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e849bd08bd364ecfb73d149a413c40eb : SUCCESS in 7m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/992852fb208443a8ae266314f454eca7 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75f7a46745ff43d8b91b49adcb8ab652 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a462e5d8329d4f9b8246110049ea3e84 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afc6e9506e09421bb5c62f54239c3de1 : SUCCESS in 7m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef662760c569465ca69e3758f379c602 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bccc3e5fd3e4275a58e1d228d8cfb78 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89e20ab5d3bf498cbf2a0389ac922ac9 : SUCCESS in 6m 35s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/6e36e03c4db241e09ca9136b8df72ffa : SUCCESS in 1h 03m 05s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/7b02361ad64a41178cb6e9ec8e68bf93 : SUCCESS in 50m 25s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ff585c4f694440b59699f4ddd1d7ec33 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":3},{"id":"4165f81d2610c4f294c3c7f7fe2bf8ddc55e3afb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 14:16:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8924fe4aa2905d00695e7c0ec17f9d052458bf50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 14:19:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0e0889839d8a44789d86d97350506b2c : SUCCESS in 2m 33s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f66d1b1b951c466f9695640a30fda9fb : SUCCESS in 2m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/158777a1c75946d78364654f3664f837 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3fe89808c60ffa1392ff82d25988d452bde2166a","revisions":{"2bd16df7942c33384563f70a11cb0a29df1512cf":{"kind":"REWORK","_number":1,"created":"2020-04-02 06:11:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/49/716849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/49/716849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/716849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/716849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/716849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/49/716849/1"}}},"commit":{"parents":[{"commit":"977b5ef7cecbd6254365a06bf8bec7287a3ca2ba","subject":"Add entrypoint to container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/977b5ef7cecbd6254365a06bf8bec7287a3ca2ba"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-02 05:51:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-02 06:10:45.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nUpdate local hacking checks for new flake8.\n\nRemove hacking and friends from lower-constraints, they are not needed\nto be installed at run-time.\n\nChange-Id: I1c5f5a85af02d9f11e475b6088fe1d9d9d252083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2bd16df7942c33384563f70a11cb0a29df1512cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2bd16df7942c33384563f70a11cb0a29df1512cf"}]},"branch":"refs/heads/master"},"c5e0e4a5e48516ee6ebd5d4ff1c9f580e5fe543f":{"kind":"REWORK","_number":2,"created":"2020-04-02 08:22:41.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/49/716849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/49/716849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/716849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/716849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/716849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/49/716849/2"}}},"commit":{"parents":[{"commit":"977b5ef7cecbd6254365a06bf8bec7287a3ca2ba","subject":"Add entrypoint to container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/977b5ef7cecbd6254365a06bf8bec7287a3ca2ba"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-02 05:51:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-02 08:22:35.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nUpdate local hacking checks for new flake8.\n\nRemove hacking and friends from lower-constraints, they are not needed\nto be installed at run-time.\n\nChange-Id: I1c5f5a85af02d9f11e475b6088fe1d9d9d252083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c5e0e4a5e48516ee6ebd5d4ff1c9f580e5fe543f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c5e0e4a5e48516ee6ebd5d4ff1c9f580e5fe543f"}]},"branch":"refs/heads/master"},"3fe89808c60ffa1392ff82d25988d452bde2166a":{"kind":"REWORK","_number":3,"created":"2020-04-02 08:28:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/49/716849/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/49/716849/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/716849/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/716849/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/716849/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/49/716849/3"}}},"commit":{"parents":[{"commit":"977b5ef7cecbd6254365a06bf8bec7287a3ca2ba","subject":"Add entrypoint to container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/977b5ef7cecbd6254365a06bf8bec7287a3ca2ba"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-02 05:51:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-02 08:28:00.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nUpdate local hacking checks for new flake8.\n\nRemove hacking and friends from lower-constraints, they are not needed\nto be installed at run-time.\n\nFix flake8wrap to take arguments from tox.ini and not look at .tox\ndirectory.\n\nChange-Id: I1c5f5a85af02d9f11e475b6088fe1d9d9d252083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3fe89808c60ffa1392ff82d25988d452bde2166a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3fe89808c60ffa1392ff82d25988d452bde2166a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
