)]}'
{"id":"openstack%2Fzun~726036","triplet_id":"openstack%2Fzun~master~Ib33e812cacf96c560fce32c0af43d74ac1074ee1","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ib33e812cacf96c560fce32c0af43d74ac1074ee1","subject":"Monkey patch original current_thread _active","status":"MERGED","created":"2020-05-07 03:25:58.000000000","updated":"2020-05-15 02:31:14.000000000","submitted":"2020-05-15 02:29:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726036-1589509785640-299f2a5a","meta_rev_id":"0245277964d346ceca5e003a0f20e3ebbf8d08ad","_number":726036,"virtual_id_number":726036,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 02:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":2,"date":"2020-05-15 01:24:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":1,"date":"2020-05-15 01:24:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 01:24:58.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2020-05-15 02:29:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3083157aaa5a80bb140dfae080ce0001bc154dc7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-07 03:25:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a90d16e63bfc9c80a18a7d6e36f5464efd8058e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 04:35:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47ebe67a5a654e7e8199a8e8edb44f2e : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46374c24b30445c597d0a2058e889312 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373be7940ff24a1f8b8ddd4d31ea0bcc : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1418fdfebe84e18a6fa9dabe18e5ecc : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e977e6280e224f0fa06731d5a5f91f9d : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/434c96fea2744f8ebcf9456b3e54d01c : SUCCESS in 5m 58s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d7cc06abd1cd4311b2319ba20da07031 : SUCCESS in 47m 45s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/1c878a8a81084fa4beb128384de2eb00 : SUCCESS in 46m 54s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/5a535f3eb767485696aba3c818466bae : SUCCESS in 1h 04m 18s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/741d15b9b8f248aab92433db4ff254be : SUCCESS in 34m 54s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/e6084e902f8740daa96543007c818b53 : SUCCESS in 44m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca3605df8c8153b993130a48dca4ba43afd9be40","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2020-05-15 01:24:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"103b10c3616dc5253975effb581c592c3f6d0502","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 01:25:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89bb59936155f4d433902c2b927d16d0c4991b07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 02:29:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37e252cae426450a844a37d405d9c1bc : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9cbe4a14eda45e18ac075af6577ad81 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/623761d64b554611912b520a94c30bfb : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31b21d7b13764f1ba60c01ea36ed54d6 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad2cc292b13415b9e808f2ed1f287f3 : SUCCESS in 5m 36s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/3d7e22bf105047479f5f91266f0c9ffa : SUCCESS in 41m 46s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/248313987d514632a1af868883be7d58 : SUCCESS in 47m 38s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/03c29393d3204ef6bcba620b8f6e3354 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e85600f048b89162ab6a700fdf22f766d3bcef78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 02:29:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0245277964d346ceca5e003a0f20e3ebbf8d08ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 02:31:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c23d4f8b2d0d460c8ac681d02d74b9fd : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68dd0e8620061ac1af95e20af250bb0349254410","revisions":{"68dd0e8620061ac1af95e20af250bb0349254410":{"kind":"REWORK","_number":1,"created":"2020-05-07 03:25:58.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/36/726036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/36/726036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/36/726036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/36/726036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/36/726036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/36/726036/1"}}},"commit":{"parents":[{"commit":"7130d65e9ec8916f2a4bffd1e0e49970f4a3e8dc","subject":"Merge \"Bump default tox env from py37 to py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7130d65e9ec8916f2a4bffd1e0e49970f4a3e8dc"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-05-07 03:25:39.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-05-07 03:25:39.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: Ib33e812cacf96c560fce32c0af43d74ac1074ee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/68dd0e8620061ac1af95e20af250bb0349254410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/68dd0e8620061ac1af95e20af250bb0349254410"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
