)]}'
{"id":"openstack%2Fzun~726894","triplet_id":"openstack%2Fzun~master~I8086b6fd925d3b48b53b3165e3802d9d94977b5f","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I8086b6fd925d3b48b53b3165e3802d9d94977b5f","subject":"Remove the unused coding style modules","status":"MERGED","created":"2020-05-11 16:10:10.000000000","updated":"2020-05-17 03:54:10.000000000","submitted":"2020-05-17 03:52:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726894-1589687557755-ed83af0c","meta_rev_id":"306b7c9876af0894ae6e41467c72d2f6289540ce","_number":726894,"virtual_id_number":726894,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"date":"2020-05-17 02:57:00.000000000","_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-17 03:52:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-05-16 13:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-05-16 11:35:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-15 16:43:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-05-16 13:15:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 16:43:41.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-05-16 11:35:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-16 13:15:34.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-05-17 03:52:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4be3fffbdabf11f5eecd255b1d87b302ac1e3120","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-11 16:10:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4b38cc458d6e038d5ef23806955dd9301722fd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 17:40:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/958789fbeef84283a128c3989dcc9f57 : SUCCESS in 6m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/782acbf0b5b74c2dba988f9b656313db : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e9dcb13b0eb4905aae9412e146d06e3 : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/501e4a4de1f043ee8ba3e24e46bdbbd7 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb133b2e5e564071bff28517c6178e62 : FAILURE in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb5d83292b384848a9b1999531e8c55e : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8795154270e04d24979ae414ffcf6e8d : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb79507e142546f7a024fa61b50ffce2 : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc2012e90fd14d5c8561913c94dbcf7c : SUCCESS in 6m 08s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7728fd94b93f45ff84271aadd7d38c15 : SUCCESS in 52m 22s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/ee9f836b15a142a993423381ce2378b0 : SUCCESS in 1h 00m 37s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d98fda832dd84c6190ad1eb6063be3e4 : SUCCESS in 1h 26m 34s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/2a4e2560d81846208590b5490a857f97 : SUCCESS in 35m 57s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/cd2d31582d5b43f28b8b4713059e235c : SUCCESS in 50m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1037619694034482d02c787465601ebea1934f24","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-12 01:49:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9ed15709b0ef6fa8df1369255597f0545a847618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 02:57:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4b5e8b2f1340485e9c7cd6c957df1a82 : SUCCESS in 4m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8a120894d3544300a25e344949dab8f6 : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1563e09e0b0046d48f9c915e28b0e166 : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87e3ee010a014388adfad84a940fe358 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dbaea45949941e3a4891197709f0e94 : FAILURE in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d04be68259e043818a68c2c6386ba49f : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ab17bb503d84c708f437dd76cb6205b : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593e1c56576c43caa8aa78d36e538818 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9645dced657a44a4b60c1e404ef77e0e : SUCCESS in 4m 13s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/37ea9a43fafc4ef2a4cc8d62d52de85c : SUCCESS in 55m 01s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/3b2ee7aa1a9447e4bdd538f1eb1eb7a4 : SUCCESS in 46m 55s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/8ff62ef8150d4742ac28fed1141ae4b8 : SUCCESS in 1h 03m 03s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/a45330440af54f8580c30a690c617418 : SUCCESS in 36m 06s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/effc06d7ff6c4b2c8cc0384697fcb8c4 : SUCCESS in 45m 34s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/container/docker/driver.py","accounts_in_message":[],"_revision_number":2},{"id":"de7ace040e34a9704934d3aaba8761bd8ee745b8","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-12 08:15:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"829f844395251e899a3b6add8581314558f7a329","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-12 08:16:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"94705579a075ad0dc27d5bcd28e81564a52b3bba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:28:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c2338db73c47403bb064d69c73504f6e : SUCCESS in 5m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/50b14fb4771e44759a111da62ba7bca9 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0ccbf21109749dbb3d546db4642bb2f : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7bc5d1c2a9c47c9828e6106258efc76 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6369a6da5638438b925dc5ca1dd7cafa : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f8e8b7e7f664ca097682bc322ed313d : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27980e89f80049828b31541c1838c464 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53978027cf174eddaa6fed592c484898 : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1a8d37c02b0489a893fc8ad72777220 : SUCCESS in 4m 50s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/1e062ac1516c40e28796332831f6fcbc : SUCCESS in 47m 07s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/fbd0187df455439a8692121bdd09098f : SUCCESS in 49m 36s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/dbadc755e6ec4a0eb262b53120faffb8 : SUCCESS in 1h 05m 29s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/d180dea98072421190b759702e45930c : SUCCESS in 33m 09s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/14e4bd1916444e779ce109b8bd225491 : SUCCESS in 43m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0abf2a73825b2da424a538340e3f1ef3dbc7ce49","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-12 13:20:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"494d9c49c2b143ad7f0ae9fb9945b1d2ad3873b8","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-12 15:02:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"47d77bc2515029768801187ee4e023505b853376","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-12 15:03:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"60082165510fc850b9a2b539d67adbb0687a92f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 16:07:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/edca82ff1a8b4d1a842222e520f070bd : SUCCESS in 4m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c06c548b397465f8d2bf54e3467a435 : SUCCESS in 3m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f3ca3c2171147f5aaf8ccaa91d45f68 : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4db5c58c94c84531ab4c3610b234dc33 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5da8a6e45d684e2d8728005bcb7fbead : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c16781e2abfa4649bea1f4f3a14d9561 : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c14a2ac6fa34a14bcf2ad54c4ccc71f : SUCCESS in 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/959e74da6d6742bfa55b8c5c519e42de : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef9ab247b9dc44ddaf811db1c94b598b : SUCCESS in 4m 04s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/cca0d769e887419d96cd15ffda714fa6 : SUCCESS in 54m 39s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/6c0b5e60425b425788a5568dd84c9f0b : SUCCESS in 50m 08s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d0b332741cb14f33832b91e0cb21dca4 : SUCCESS in 1h 02m 17s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/566dbf5a6c5c4211957bdd5175d7ce6d : SUCCESS in 41m 47s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/314d3ba07f67476799cef9c3df9315c4 : SUCCESS in 55m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1662b6167cef956a8231146c78ee4b4856f75dcd","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-13 14:52:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5f451f1fa8efe09ecfa5c5d4028d39b4e142e15b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:07:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9d978c1e32ef421796cd382b90e9cdcb : SUCCESS in 7m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/60ee6ef09ba449d69b9272b7ac6867bc : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3455c4403cab4020a996315e8f0f0ba6 : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86e47fedd37142e2b5a763871d2fe579 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a34627049f5a453ab7817b7be3fb1b68 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82f404b1525c4c9c84fa17ce96483b12 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b97b4f66169848b5a4161fb7c45f6688 : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af5910ef6ff48dc8aa24294dfe463e0 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/533e2609843945f78bd79f2f3321a480 : SUCCESS in 4m 36s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/697e057835c74eb4b49f64c9bd2978d6 : FAILURE in 19m 55s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/b2d6b72abc7145e3b2c5001411b848b6 : SUCCESS in 49m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/61af1ce3d35b42f7bf05d52d970e7618 : SUCCESS in 1h 08m 52s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/d4174c1cecdf4eb5a507d1dcbf362c04 : SUCCESS in 41m 10s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/242a8919f8b04baeb89e5dfb5b9d744d : SUCCESS in 55m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"670c3fcb451d8a7a3ad052f02a601d5b2f6d5e25","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-13 16:08:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ae6160cfc64b220f560c467bfae04a127bf72ab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 17:57:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8ca91eac99df4ce5bf9e4201239f5810 : SUCCESS in 7m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/951f72fcad0241a3bd4132e8da95e4c4 : SUCCESS in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08b900061fd14ea1bf153aaa2c358133 : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db8037e928ed4bc3a195ac20c327f5ce : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecaa15541fbe44daa52957a1bf033df9 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c39f6dd448ca4f999030aca5709d111b : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66c3546099464dc5a5dff839a2ce5512 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffc9f70b7b4b415a8977703235df2b70 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40ea1dd2154e495a917a1ae40e29b028 : SUCCESS in 3m 57s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/79c5ec60d6f249a7a892a25cce40cfdd : SUCCESS in 51m 32s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/d5a491c7374846fc96903b88b95772ce : SUCCESS in 49m 41s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/4e5d2c15c1a7426697c630c0eaf1eb76 : SUCCESS in 1h 12m 20s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/1b35d99f3194484b945b64dfc946af74 : SUCCESS in 36m 56s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/76fb50fdbce946269aa51819cb941573 : SUCCESS in 52m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f305e62d5bf1126eb00fc544bb46667ae35edfd2","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-15 16:43:41.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"597ade27fb425fae3b725a07a2f233bac4c2d76e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-16 02:56:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"37558e0f0779639fdaa5a73a33eb637e2fafa83e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 02:56:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6c21a25070e7cdf6d850b0a59eca155a5d93888f","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-16 03:08:15.000000000","message":"Patch Set 5: -Code-Review Workflow-1\n\nActually, let me consult the requirements team for this.","accounts_in_message":[],"_revision_number":5},{"id":"7bc99f099621f672861fc01d9c2b56206cf3676e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-16 03:14:49.000000000","message":"Patch Set 5:\n\nHi Sean,\n\nI am not sure about this patch. Would you provide opinion from requirements team perspective. I mainly want to know whether it is correct to drop things from lower constraints like this?","accounts_in_message":[],"_revision_number":5},{"id":"4ad6d8606cf40e1f43ef3fb94688d6094a186604","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 03:52:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/947f2a28aab743e0a7ff2a94f29cabdd : SUCCESS in 4m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6de2b463720e47329958767fedfbfd31 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36a3a7611a724495a026177e16e5a816 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cea1b5eccb542099051428c6f2cbc89 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66c9e25c3f6546cc8206fab2099b35b6 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/960ed624c35a4dc7805baa8fec1466f3 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eea9bb5515f4b5bb549c5bea622fdbd : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/710b519682374239a8961fe4e74dde1c : SUCCESS in 4m 00s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ab16c8aa97ce494aadad3a8fc2c12e25 : SUCCESS in 50m 40s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/fa1807376bad400e8569eb3df5e30521 : SUCCESS in 47m 12s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/1b8da744787a43eca494a3e0404d2574 : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":5},{"id":"1d5f7953a743902272d6446bb230e637f2dc7266","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-16 11:35:36.000000000","message":"Patch Set 5: Code-Review+1\n\n\u003e Hi Sean,\n \u003e \n \u003e I am not sure about this patch. Would you provide opinion from\n \u003e requirements team perspective. I mainly want to know whether it is\n \u003e correct to drop things from lower constraints like this?\n\nThanks for checking. Yes, that should be fine. Testing packages can be constrained, but it is not necessary.\n\nThese look good to remove.","accounts_in_message":[],"_revision_number":5},{"id":"9802a2515de55cbbf6b0ea3068ade6f53f102399","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-16 13:15:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n\u003e \u003e Hi Sean,\n \u003e \u003e\n \u003e \u003e I am not sure about this patch. Would you provide opinion from\n \u003e \u003e requirements team perspective. I mainly want to know whether it\n \u003e is\n \u003e \u003e correct to drop things from lower constraints like this?\n \u003e \n \u003e Thanks for checking. Yes, that should be fine. Testing packages can\n \u003e be constrained, but it is not necessary.\n \u003e \n \u003e These look good to remove.\n\nThanks Sean.","accounts_in_message":[],"_revision_number":5},{"id":"3ddf5b8990b68fd2804abfaf921357faf65c0d34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 13:15:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cb949db08fb9d35d7003530bf3d491fd50eb61d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 14:15:19.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5778bcaed21a4025a60d60125e68caa4 : SUCCESS in 5m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0dcab2cdc01448ff88da935818f303e7 : FAILURE in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7c88f62b88f4248aaa234c80393ee88 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47496c49a150486ab2f6ac2d1bd7a228 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2b1634f84f24e38bbe45d17a26d668d : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baa8154172f944a6adac7af6e33ca2ba : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41e550ee357e458eba1a20dcbbf4b599 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5617a428b6314f5c922baa2f2abfbbd0 : SUCCESS in 4m 51s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/367cfe8234444bfb944bf9600fb5525e : SUCCESS in 46m 07s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/880988cca8984b7f94ed76feb8e844db : SUCCESS in 49m 51s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/a5befacab6104aed8140643d979a2541 : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":5},{"id":"98cb24560d6c8c2acbfc54af6fc0778747fb3cc9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-17 02:01:14.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8f1e62cff0490b28f1ee9a18c24831e826aeb710","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-17 02:57:00.000000000","message":"Patch Set 5:\n\n\u003e \u003e \u003e Hi Sean,\n \u003e \u003e \u003e\n \u003e \u003e \u003e I am not sure about this patch. Would you provide opinion from\n \u003e \u003e \u003e requirements team perspective. I mainly want to know whether it\n \u003e \u003e is\n \u003e \u003e \u003e correct to drop things from lower constraints like this?\n \u003e \u003e\n \u003e \u003e Thanks for checking. Yes, that should be fine. Testing packages\n \u003e can\n \u003e \u003e be constrained, but it is not necessary.\n \u003e \u003e\n \u003e \u003e These look good to remove.\n \u003e \n \u003e Thanks Sean.\n\nthanks","accounts_in_message":[],"_revision_number":5},{"id":"ccaa30ac25812b163aee422a10f84698b0b81a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 02:57:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/40638a3b0d2c48329990d938edae2e66 : SUCCESS in 4m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c5cb4b92959d4ddd9eadaae497c64c73 : SUCCESS in 4m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65252d6b88ee43678cd047d955925c91 : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63f987b1d00e4531811b9646e305b5c3 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8700576cbed546c49f0b40371fe804de : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9efbdb95d31b457888914a8111ea4fcf : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d7ec7f2dbf74dd982dfa14489637a29 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696acc15bc3d4ea9b06eeaab3282946c : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba27ecc773b64064841052668b9cd813 : SUCCESS in 3m 19s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/000d53b24df94bba876b7585b9c2b00b : SUCCESS in 42m 52s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/524a8e78b618465cb5f9e32ef56534e3 : SUCCESS in 41m 43s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ae1183a75327487ea8802fad68be50f5 : SUCCESS in 55m 36s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/8c0c4ee963bf4959abdc8b7d1ac00749 : SUCCESS in 31m 00s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/a18588531d3a462895077ed3106ceb00 : SUCCESS in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"97605b76e75fbfee84c4d6653384b3740d8e7b17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 02:58:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b8e0b281bbb402c1e52b56d24abda9b51df047b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 03:52:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/50c1c22ab8744415abc725f7869554f6 : SUCCESS in 4m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/13545f989263412b8dc3f09e476992f4 : SUCCESS in 3m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfbe665d98444ca58f14c81365529285 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe3c445e25064b1c875d2aca749bfd02 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71aae827c2f448239e32c1c36331ef3a : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cf121b66c94475cb35ed9b8906d3b47 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ad1b25879547ef9cbc65eed446ff50 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c776e5389f14bedb9bb2fbc986ff42b : SUCCESS in 4m 33s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/c3482a5ed2e8476fbb7db923fefd1d17 : SUCCESS in 48m 18s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/16aae47d10c54e34955372165abecd47 : SUCCESS in 44m 38s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d232c11a6e294d4d846d5d5ce20d6ea0 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":5},{"id":"126248f955c4296b36456ad0cd1938fa8e1f7717","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 03:52:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"306b7c9876af0894ae6e41467c72d2f6289540ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 03:54:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b4c47980d77b461eacf2b844d876caa4 : SUCCESS in 58s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088d2d0c211e4cf19c7044c650967426 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd998daaad6747b48cf9cbc573b17bbe : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"01767f36fef4c95764817fc79a1d8e1c9578d4d8","revisions":{"e099c8dc6fca26941e125a7e0e56e506a4d460ae":{"kind":"REWORK","_number":1,"created":"2020-05-11 16:10:10.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/94/726894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/94/726894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/94/726894/1"}}},"commit":{"parents":[{"commit":"397eca92325461f2ccfcf3b8be4a04ced56fc5c8","subject":"[Part2]: Remove the six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/397eca92325461f2ccfcf3b8be4a04ced56fc5c8"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-11 16:07:58.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-11 16:07:58.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nPython modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\nChange-Id: I8086b6fd925d3b48b53b3165e3802d9d94977b5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e099c8dc6fca26941e125a7e0e56e506a4d460ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e099c8dc6fca26941e125a7e0e56e506a4d460ae"}]},"branch":"refs/heads/master"},"1f8b9e2469a660739334bbfcedead8a92e0c8bf9":{"kind":"REWORK","_number":2,"created":"2020-05-12 01:49:29.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/94/726894/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/94/726894/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/94/726894/2"}}},"commit":{"parents":[{"commit":"397eca92325461f2ccfcf3b8be4a04ced56fc5c8","subject":"[Part2]: Remove the six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/397eca92325461f2ccfcf3b8be4a04ced56fc5c8"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-11 16:07:58.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-12 01:48:49.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\n1. Python modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\n2. remove mock\n\nChange-Id: I8086b6fd925d3b48b53b3165e3802d9d94977b5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1f8b9e2469a660739334bbfcedead8a92e0c8bf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1f8b9e2469a660739334bbfcedead8a92e0c8bf9"}]},"branch":"refs/heads/master"},"7545c0a5ee7ad86eb832da50aac89b817578371f":{"kind":"REWORK","_number":3,"created":"2020-05-12 08:15:36.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/94/726894/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/94/726894/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/94/726894/3"}}},"commit":{"parents":[{"commit":"397eca92325461f2ccfcf3b8be4a04ced56fc5c8","subject":"[Part2]: Remove the six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/397eca92325461f2ccfcf3b8be4a04ced56fc5c8"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-11 16:07:58.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-12 08:15:11.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\n1. Python modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\n2. remove mock\n\nChange-Id: I8086b6fd925d3b48b53b3165e3802d9d94977b5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7545c0a5ee7ad86eb832da50aac89b817578371f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7545c0a5ee7ad86eb832da50aac89b817578371f"}]},"branch":"refs/heads/master"},"9d76546705c21f795c0cbd9228a7932da5bd1020":{"kind":"REWORK","_number":4,"created":"2020-05-12 15:02:12.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/94/726894/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/94/726894/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/94/726894/4"}}},"commit":{"parents":[{"commit":"397eca92325461f2ccfcf3b8be4a04ced56fc5c8","subject":"[Part2]: Remove the six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/397eca92325461f2ccfcf3b8be4a04ced56fc5c8"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-11 16:07:58.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-12 15:01:29.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\n1. Python modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\n2. remove mock\n\nChange-Id: I8086b6fd925d3b48b53b3165e3802d9d94977b5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9d76546705c21f795c0cbd9228a7932da5bd1020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9d76546705c21f795c0cbd9228a7932da5bd1020"}]},"branch":"refs/heads/master"},"01767f36fef4c95764817fc79a1d8e1c9578d4d8":{"kind":"REWORK","_number":5,"created":"2020-05-13 14:52:07.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/94/726894/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/94/726894/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/726894/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/94/726894/5"}}},"commit":{"parents":[{"commit":"8dd2b79cbcfed019801b144db1d110225500fad4","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8dd2b79cbcfed019801b144db1d110225500fad4"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-11 16:07:58.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-13 14:51:08.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\n1. Python modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\n2. remove mock\n\nChange-Id: I8086b6fd925d3b48b53b3165e3802d9d94977b5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/01767f36fef4c95764817fc79a1d8e1c9578d4d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/01767f36fef4c95764817fc79a1d8e1c9578d4d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
