)]}'
{"id":"openstack%2Fzun~729432","triplet_id":"openstack%2Fzun~master~I5d01e8946fc722627759b10598d867adc6dd2c9c","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I5d01e8946fc722627759b10598d867adc6dd2c9c","subject":"Fix pygments style","status":"ABANDONED","created":"2020-05-20 00:01:18.000000000","updated":"2020-05-22 01:10:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"042e3bfb68c705b3c38581d033c22d38595bf8b4","_number":729432,"virtual_id_number":729432,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-21 02:22:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 02:22:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42f9178a6e041d2306d1ed4c1e644f0dbff9b3a5","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-20 00:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdf2cfecfde3039b46928b3867941e92621126d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 01:17:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1c080b7cf82f4ed8a4ddaff98c264f36 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473fb82e7df848938a04b79d36e7d354 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5639b3f9b9634a9981ff0552a75e0962 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce05301dbf5147ea920c6c33b7975ae7 : SUCCESS in 4m 50s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/191a7520d7794b529921ea5145f20699 : SUCCESS in 52m 47s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/f7d83ce1b4014085aeb29530581dddf4 : SUCCESS in 52m 00s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/062834e43a004039aee65101f44a9a5f : SUCCESS in 1h 14m 56s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/54a11d9320bc4257b6cc8aa9dcae6232 : SUCCESS in 32m 34s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/3ca21f8a5bd74c07a4e9a98e49793715 : SUCCESS in 46m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccc4272b597e8529148a3fa71eae5bd07fa0c91e","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-21 01:19:07.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"55a7a2913ce252e8b50a78e148e3a23205aedb5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 02:22:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b1f873c5e72f4257852071dbf6c988f9 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f754793daeab47d380c18948b85532ef : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4701abcddf47c6b08402ee614ced8e : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63979006fc004ff28c9b9763cc81184c : SUCCESS in 3m 13s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/3c662c2512a543a1b7019bf8344cc1eb : SUCCESS in 48m 10s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/4b47b9bc1f4a420aa4587927fd02cb54 : SUCCESS in 45m 35s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/af2748873a36448998bf78d1680e1ef2 : SUCCESS in 58m 30s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/6f0b66af8ecd45e9b44f214bcef672ff : SUCCESS in 40m 41s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/81de13b38db64064ad113d857bffe5bb : SUCCESS in 43m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"042e3bfb68c705b3c38581d033c22d38595bf8b4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-22 01:10:24.000000000","message":"Abandoned\n\nWe merged another patch: https://review.opendev.org/#/c/729869/ . This patch is not needed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"edf8b738ed6722c75b931747177802a891ef3320","revisions":{"efda5e5c930cec4f0f492608fa3e8d4bdf6a57a1":{"kind":"REWORK","_number":1,"created":"2020-05-20 00:01:18.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/32/729432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/32/729432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/32/729432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/32/729432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/32/729432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/32/729432/1"}}},"commit":{"parents":[{"commit":"42b7fb8b95edd0db601d92d0abd6c56f6f33902f","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/42b7fb8b95edd0db601d92d0abd6c56f6f33902f"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 00:00:52.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 00:00:52.000000000","tz":480},"subject":"Fix pygments style","message":"Fix pygments style\n\nNew theme of docs (Victoria+) respects pygments_style.\nSince we starts using Victoria reqs while being on Ussuri,\nthis patch ensures proper rendering both in Ussuri and Victoria.\n\nChange-Id: I5d01e8946fc722627759b10598d867adc6dd2c9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/efda5e5c930cec4f0f492608fa3e8d4bdf6a57a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/efda5e5c930cec4f0f492608fa3e8d4bdf6a57a1"}]},"branch":"refs/heads/master"},"edf8b738ed6722c75b931747177802a891ef3320":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-21 01:19:07.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/32/729432/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/32/729432/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/32/729432/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/32/729432/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/32/729432/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/32/729432/2"}}},"commit":{"parents":[{"commit":"42b7fb8b95edd0db601d92d0abd6c56f6f33902f","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/42b7fb8b95edd0db601d92d0abd6c56f6f33902f"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 00:00:52.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-21 01:19:05.000000000","tz":0},"subject":"Fix pygments style","message":"Fix pygments style\n\nNew theme of docs respects pygments_style.\n\nmore info: http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\n\nChange-Id: I5d01e8946fc722627759b10598d867adc6dd2c9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/edf8b738ed6722c75b931747177802a891ef3320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/edf8b738ed6722c75b931747177802a891ef3320"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
