)]}'
{"id":"openstack%2Fzun~730494","triplet_id":"openstack%2Fzun~master~I6bda2007e9ad4374e84ee86ef7f7c93a4125d862","project":"openstack/zun","branch":"master","topic":"bug/1880358","hashtags":[],"change_id":"I6bda2007e9ad4374e84ee86ef7f7c93a4125d862","subject":"Fix capsule\u0027s container status","status":"MERGED","created":"2020-05-24 04:43:39.000000000","updated":"2020-05-24 16:11:05.000000000","submitted":"2020-05-24 15:02:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730494-1590332520820-48d602ed","meta_rev_id":"d40ca1a0b94d5e1fbe4fe8d5d575dfa8c99c0812","_number":730494,"virtual_id_number":730494,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-24 15:02:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-05-24 14:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-05-24 14:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-24 15:02:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"658b24d3d744bbaf7ad1b2a302f3620cca1c1e55","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-24 04:43:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4c0b3800c7d2df70bc7dd5cee4f5e73b9b65aa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 05:48:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bc029a48e7d4962a94f7f3757719160 : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f73a73ddd6a543d8a829cf717a158b21 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f01c95d7845346d2b7229f0ebdb3ef0c : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f25776b52d94cb5ae08de4a00781256 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51e1ebe4218446549c992d3e89adacaf : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d2ff129a42b4af5a9c284d81412fc3a : SUCCESS in 6m 57s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/f830158365054e9b9ef372ac08d1ca03 : SUCCESS in 46m 58s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/56ee44c07bbb49d8bd3c66ef2d83cba1 : SUCCESS in 45m 28s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f349bfbe5a86425c89b476234bc89f29 : SUCCESS in 1h 02m 50s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/777f3e55be524a7d9c61c92d493f4cf9 : SUCCESS in 31m 32s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/c0348338c052471988b46442902b4708 : SUCCESS in 39m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6376f41d190db070cc51b7fcbebb71b3d4334be3","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-24 14:00:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c166dfce1131c9dfae1aaf3a42a6e1c8c901fc29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 14:01:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2526eaacfbeb4f9b49b083d34cbc2d4717f49808","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 15:02:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3869a5bf63e5475dbf7bfc1642687b18 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9632875534d482d8080966fb741db42 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/923fd5ba34aa4c1bbac858c6e4a4fac1 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8994347ca03e420c8b4450448c5892aa : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eadc914e966f4dbaa63913904c90557d : SUCCESS in 6m 04s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/679a6297ac4d4cf4aaf80b6a5618bcfa : SUCCESS in 47m 09s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/fa40c5033738453386ad2e681adbc399 : SUCCESS in 47m 08s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/fa5bb259582945bfa783c6d0eb894297 : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8fdeab0c2cf4467928f411c669c0354d53c45b95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 15:02:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"79ebd026435b5d4003a006600b374dedb4466609","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 15:03:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544d45031fd0496c8ab7ac1570357d03 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1},{"id":"7950e74e4657586e84d4c69700d2821f9a33fbb9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-24 16:11:05.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 24f611080435d3458fe55dfdaaa07b3d30c439ba","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef6b875f393744388838b2a8ad3dbd7e660d8e62","revisions":{"ef6b875f393744388838b2a8ad3dbd7e660d8e62":{"kind":"REWORK","_number":1,"created":"2020-05-24 04:43:39.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/94/730494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/94/730494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/730494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/730494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/730494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/94/730494/1"}}},"commit":{"parents":[{"commit":"fd28742be0ca722ad2749987bfb79c6032bfe406","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/fd28742be0ca722ad2749987bfb79c6032bfe406"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-05-24 04:42:39.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-05-24 04:43:38.000000000","tz":0},"subject":"Fix capsule\u0027s container status","message":"Fix capsule\u0027s container status\n\nSave capsule containers/init containers status to db on creation\n\nChange-Id: I6bda2007e9ad4374e84ee86ef7f7c93a4125d862\nCloses-Bug: #1880358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ef6b875f393744388838b2a8ad3dbd7e660d8e62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ef6b875f393744388838b2a8ad3dbd7e660d8e62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
