)]}'
{"id":"openstack%2Fzun~730539","triplet_id":"openstack%2Fzun~stable%2Fstein~I537bbeda79b91ca83f97cf484078a93ff5c1d25f","project":"openstack/zun","branch":"stable/stein","hashtags":[],"change_id":"I537bbeda79b91ca83f97cf484078a93ff5c1d25f","subject":"[DO NOT MERGE] Test CI","status":"ABANDONED","created":"2020-05-25 01:40:26.000000000","updated":"2020-05-25 02:57:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"73e22cc71886f4bf94ca715bd878bf8a6018cb43","_number":730539,"virtual_id_number":730539,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-25 02:40:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 02:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a52337685462b2505b6039cf32ee7ff92423d32","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-25 01:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0514b4db4bbe9b36a450d2dcddee8b7b073ec75a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 02:40:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2a4f86a64a6f4309be19239ce2cff02b : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/627e53149bc7407d9838a6291e871af0 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2caa656a4484b0da36bec86fb1d4769 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86888334b1d424881aebeb44f4121c1 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7a0dd234b0f43e8b3dddf39edce2903 : SUCCESS in 6m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a26ea21e0d0f41748828c75954f33a90 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b509950e78ee47fe9017a39c869f33c5 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8df022ed84bf46bcab0b042523823045 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59dbe2ea71434784b5ed8b70483b656a : SUCCESS in 5m 12s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/d9fa84508a84413f8b4ea1f885104f96 : FAILURE in 18m 27s\n- zun-tempest-py35-docker-sql https://zuul.opendev.org/t/openstack/build/05391c417b7e4aa880acd4f1675a3651 : SUCCESS in 46m 58s\n- zun-tempest-docker-etcd https://zuul.opendev.org/t/openstack/build/fc00b41f3368428e810067fe0b46d698 : SUCCESS in 48m 03s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/1515b33ce8db4741bf13110ae8a6b11b : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"73e22cc71886f4bf94ca715bd878bf8a6018cb43","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-25 02:57:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75673ceeb37a7fba5e4c55fa5831554b462966d3","revisions":{"75673ceeb37a7fba5e4c55fa5831554b462966d3":{"kind":"REWORK","_number":1,"created":"2020-05-25 01:40:26.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/39/730539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/39/730539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/39/730539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/39/730539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/39/730539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/39/730539/1"}}},"commit":{"parents":[{"commit":"3f4ea79e2d9fa56b9d1ae3474be19ebab5a52d68","subject":"Replace tox_envlist \u0027all-plugin\u0027 with \u0027all\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3f4ea79e2d9fa56b9d1ae3474be19ebab5a52d68"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-05-25 01:40:03.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-05-25 01:40:25.000000000","tz":0},"subject":"[DO NOT MERGE] Test CI","message":"[DO NOT MERGE] Test CI\n\nChange-Id: I537bbeda79b91ca83f97cf484078a93ff5c1d25f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/75673ceeb37a7fba5e4c55fa5831554b462966d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/75673ceeb37a7fba5e4c55fa5831554b462966d3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
