)]}'
{"id":"openstack%2Fzun~730549","triplet_id":"openstack%2Fzun~stable%2Fussuri~Ia2a4b7befc3be9541846d8b20bd532888c4391c0","project":"openstack/zun","branch":"stable/ussuri","topic":"bug/1880358-stable/ussuri","hashtags":[],"change_id":"Ia2a4b7befc3be9541846d8b20bd532888c4391c0","subject":"Fix init container status - docker","status":"MERGED","created":"2020-05-25 02:58:21.000000000","updated":"2020-06-11 00:24:08.000000000","submitted":"2020-06-11 00:22:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730549-1591834943925-418bdb2a","meta_rev_id":"75d714c18660619a8861776e157f0197ac38f658","_number":730549,"virtual_id_number":730549,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 00:22:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-06-10 23:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-06-10 23:12:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-11 00:22:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3290d737bec2cbca3ad62d3291369322987e8aed","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-25 02:58:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6a23e9363dfb0b42b5628d5cfc924337757b10ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 03:56:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d63feb7a1874fb28749838552442364 : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bee4704cfff4fd68289b3e32929f355 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/731892a921a7466c8a4b5efd55015945 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dae3fa8eb3a54fdc9b4307ba6136b40c : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5519f88fbe14407f923f290065900156 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4906f1c13d4ba18fcda9ec07583bcf : SUCCESS in 7m 01s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/0bbc152916af4f3d9ebd066db89e1f3a : SUCCESS in 57m 03s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/01276a684d7d49faa4e51841ffab6136 : SUCCESS in 48m 46s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/888ec7bdab4e4c6891fa2076410eb7bc : SUCCESS in 57m 23s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/9f4c8725944b48d1901ba94e04c6cd63 : SUCCESS in 41m 29s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/57e15db356be4952b2cc9513348841e2 : SUCCESS in 42m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5c56383b53d39a6ff8f54f3e01679dc6fd9413d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-10 23:12:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"23386409f44fb842f95740cf64c9bd7f60860b5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 23:13:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3a1e24518d5a1181d7e04ebd5dfee5fed17ea1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 00:22:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60ebd23424634534a993e70ef46b529a : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795e82545ec7422d9a6fd26c822bce4a : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c3482b73fa14c34b7e2de0c699f61d9 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df95d4a283e3482498028acf00ebbc0c : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/516e18a12c274dc4b8d1e21210a71407 : SUCCESS in 5m 38s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/8b7d236baa11454a898dc6148e445d08 : SUCCESS in 48m 47s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/1555b0aeadbf4a9ebe13c59fbbe08648 : SUCCESS in 46m 41s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/e5fb706fb6b245158ded6ac9cf0cae41 : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d82fe5b067ff20191c6cf8e3f9e43193749b3f36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 00:22:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0a3292b22b24fdb88b8caac5e1260de0951e77c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 00:24:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9aa1b1a31a4e8e937be41c39631b5f : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f455bffc418096ad223caf27b860e07ec96535c5","revisions":{"f455bffc418096ad223caf27b860e07ec96535c5":{"kind":"REWORK","_number":1,"created":"2020-05-25 02:58:21.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/49/730549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/49/730549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/730549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/730549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/730549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/49/730549/1"}}},"commit":{"parents":[{"commit":"24f611080435d3458fe55dfdaaa07b3d30c439ba","subject":"Fix capsule\u0027s container status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/24f611080435d3458fe55dfdaaa07b3d30c439ba"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-05-24 16:17:10.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-05-25 02:58:21.000000000","tz":0},"subject":"Fix init container status - docker","message":"Fix init container status - docker\n\nSave capsule init containers status to db on creation\n\nChange-Id: Ia2a4b7befc3be9541846d8b20bd532888c4391c0\nRelated-Bug: #1880358\n(cherry picked from commit 895d211aa2cfe9643be9b803994672c0b473fbc3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f455bffc418096ad223caf27b860e07ec96535c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f455bffc418096ad223caf27b860e07ec96535c5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
