)]}'
{"id":"openstack%2Fzun~732082","triplet_id":"openstack%2Fzun~master~Ie6c1638a804899dc14b84aea1b23eba6104f482a","project":"openstack/zun","branch":"master","topic":"bug/1880202","hashtags":[],"change_id":"Ie6c1638a804899dc14b84aea1b23eba6104f482a","subject":"Remount container volume on host reboot","status":"MERGED","created":"2020-06-01 03:19:03.000000000","updated":"2020-06-07 23:57:56.000000000","submitted":"2020-06-07 23:56:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732082-1591574198299-68f2f97b","meta_rev_id":"e02e5898c9e424db174fc2c6faf1b048742e6b82","_number":732082,"virtual_id_number":732082,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-07 23:56:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-06-07 22:34:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-06-07 22:34:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-07 23:56:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2db32adbaf00cc0d7558a9beada9d769364bc55","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-01 03:19:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bef47e979edc698627f55307a32eb7e949e8d122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 04:20:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db58bb00628c4c998340929c88f0383b : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5afa9691df974e869144bf206eeb4bcf : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3777175fcf54449f90dbc30bd0f2a764 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/637154e376384ca0be8a4ac31549fcdd : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acd59c63bffe4dd8a0061b760d71cd78 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/510a4bf5575a4ae1bf43bc0fd362ee13 : SUCCESS in 8m 05s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/68f6bdef05ce4dd2b35dd929df915225 : SUCCESS in 50m 29s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/43156153163e43ecbf0a9250bdbd08de : SUCCESS in 46m 24s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/198de5be6f8848768a0f83af993b6ae6 : SUCCESS in 56m 44s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/5b616f968d9f44e2a11c21b68169a0da : SUCCESS in 32m 29s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/7327393870eb438183183e92ce0f9a42 : SUCCESS in 43m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d8cde81f036d3ab38d2911f1ea3330ef18cd579","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-01 05:03:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c4b330e8806e163ac82746c9ae916004d4a7693a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-01 05:13:28.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5f082ed49c7142fd681fff99416a14388069e6fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 06:36:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f8455e8ccbb4589941af9be151805d7 : SUCCESS in 7m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ea54353dfb640f7a7cd6e7809837ab3 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/171c7dd370f241379d3091424acd07c0 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/980805c443904b64bc5b160ced10581d : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1764153309094ecc9369a694b4100635 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa6b385bb47545a09fa49dd2a5cde6c3 : SUCCESS in 7m 23s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/204fe751e93444eea4ddde4c393a0d0b : SUCCESS in 45m 02s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/751a87b29f7c439197f4707eee27a2d9 : SUCCESS in 1h 04m 50s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2a24bf9c29ef455aa5fb3d4fbeab91b5 : SUCCESS in 1h 22m 09s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/95a29f1e48964895956275e93972cfc9 : SUCCESS in 32m 20s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/7f2dc933974f496e8e0ea478bea869b6 : SUCCESS in 57m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f7f7f751e15545b10369966310ecc7f8d3bfb8a9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-01 15:31:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6c380dcc2c01c0521c9c68a9d09d8333c6d4dee4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-01 15:32:09.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"fc282ee3ade3d04a1aa6f44e9a1d807d799b31c6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-07 17:21:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fcc9ac502d583b8f073aecd35cc39a2cfce9db4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-07 18:40:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9722758bcf2456296e86ab30f2da9b3 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f10813d2f2bd4403b1994303db25e94e : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0b8e8ef4cb64605b83079b341db6089 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6666384d96784f459c83479138657773 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6b42afc322c42ca9fd7bcc7fca424fd : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d5358acdaac40d4a1225e22d0716ae2 : SUCCESS in 7m 49s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/554d51eb017a4b77998ccb1d9cf5ddd2 : SUCCESS in 46m 59s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/28a2077c1c3d432fb30605a6f0ea0ee9 : SUCCESS in 35m 44s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/65e35e032484470d8bcadaaffe31d05e : SUCCESS in 57m 29s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/9fe91c7526ad420c8e0e1d42dc2e59e0 : SUCCESS in 31m 14s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/df56004086af45e6b54f6e0f18b73d5d : SUCCESS in 42m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8c6168dfb665ac1753742dfb69bd198a53df8c8b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-07 22:34:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9bc6a0e8a889a898017d506fc5f6140c16179a29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-07 22:35:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1421fb05325807e74e9551d3cf8a945fbed0044f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-07 23:56:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b4e57a0163f7d0db7a8506eea347a2590241665d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-07 23:56:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85b3bd3aa5074ab5b433ca394b7538ce : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81fe526eb4c943a48a3c1b5cf0633f96 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd551f03a69d4037829895aebaedaa4c : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29a4bca606374f82bcac58b83b0db688 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd311d95f13c478580d01d0612bc7db4 : SUCCESS in 5m 52s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/83f304c4cd8049a1a555e940f1aa45f7 : SUCCESS in 57m 25s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/356d0437cf704f7ab150ca666d14edb0 : SUCCESS in 39m 40s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/3db4c17165e444c7bafdf749caee6524 : SUCCESS in 1h 20m 41s","accounts_in_message":[],"_revision_number":4},{"id":"3ec17b614639b45de978989f474894a5b543e1af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-07 23:57:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/586abfff44564824a2513eb9fca5d94b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9b9bbfe6bd9011868aa8ac1c4268a531d5d019ca","revisions":{"2149625f5e677be30137f5dfacba1c1050f0a6d5":{"kind":"REWORK","_number":1,"created":"2020-06-01 03:19:03.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/82/732082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/82/732082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/82/732082/1"}}},"commit":{"parents":[{"commit":"0e691ad6d6499bcfa7bbd03030415485396ff2d1","subject":"Merge \"Fix init container status - docker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0e691ad6d6499bcfa7bbd03030415485396ff2d1"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-01 03:17:18.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-01 03:17:18.000000000","tz":0},"subject":"[WIP] Remount container volume on host reboot","message":"[WIP] Remount container volume on host reboot\n\nChange-Id: Ie6c1638a804899dc14b84aea1b23eba6104f482a\nCloses-Bug: #1880202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2149625f5e677be30137f5dfacba1c1050f0a6d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2149625f5e677be30137f5dfacba1c1050f0a6d5"}]},"branch":"refs/heads/master"},"912466b690972abf02c936ae0f98a074518df8ef":{"kind":"REWORK","_number":2,"created":"2020-06-01 05:03:00.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/82/732082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/82/732082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/82/732082/2"}}},"commit":{"parents":[{"commit":"0e691ad6d6499bcfa7bbd03030415485396ff2d1","subject":"Merge \"Fix init container status - docker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0e691ad6d6499bcfa7bbd03030415485396ff2d1"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-01 03:17:18.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-01 05:02:50.000000000","tz":0},"subject":"[WIP] Remount container volume on host reboot","message":"[WIP] Remount container volume on host reboot\n\nChange-Id: Ie6c1638a804899dc14b84aea1b23eba6104f482a\nCloses-Bug: #1880202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/912466b690972abf02c936ae0f98a074518df8ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/912466b690972abf02c936ae0f98a074518df8ef"}]},"branch":"refs/heads/master"},"209ed8ae2c634ffde59c91812a778b38af2a42e7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-06-01 05:13:28.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/82/732082/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/82/732082/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/82/732082/3"}}},"commit":{"parents":[{"commit":"0e691ad6d6499bcfa7bbd03030415485396ff2d1","subject":"Merge \"Fix init container status - docker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0e691ad6d6499bcfa7bbd03030415485396ff2d1"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-01 03:17:18.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-01 05:12:14.000000000","tz":0},"subject":"Remount container volume on host reboot","message":"Remount container volume on host reboot\n\nIf the compute host is reboot, it needs to re-connect to the\ncinder volumes and remount them into the compute host.\n\nChange-Id: Ie6c1638a804899dc14b84aea1b23eba6104f482a\nCloses-Bug: #1880202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/209ed8ae2c634ffde59c91812a778b38af2a42e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/209ed8ae2c634ffde59c91812a778b38af2a42e7"}]},"branch":"refs/heads/master"},"9b9bbfe6bd9011868aa8ac1c4268a531d5d019ca":{"kind":"REWORK","_number":4,"created":"2020-06-07 17:21:13.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/82/732082/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/82/732082/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/732082/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/82/732082/4"}}},"commit":{"parents":[{"commit":"90f120bebae1e9dcca88c1693da1eaaf2016f604","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/90f120bebae1e9dcca88c1693da1eaaf2016f604"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-01 03:17:18.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-07 17:21:13.000000000","tz":0},"subject":"Remount container volume on host reboot","message":"Remount container volume on host reboot\n\nIf the compute host is reboot, it needs to re-connect to the\ncinder volumes and remount them into the compute host.\n\nChange-Id: Ie6c1638a804899dc14b84aea1b23eba6104f482a\nCloses-Bug: #1880202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9b9bbfe6bd9011868aa8ac1c4268a531d5d019ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9b9bbfe6bd9011868aa8ac1c4268a531d5d019ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
