)]}'
{"id":"openstack%2Fzun~732981","triplet_id":"openstack%2Fzun~master~I564d6a8718feadd5f9d3cf12c73f66d13ec47d45","project":"openstack/zun","branch":"master","topic":"drop_future_imports","hashtags":[],"change_id":"I564d6a8718feadd5f9d3cf12c73f66d13ec47d45","subject":"Stop to use the __future__ module.","status":"MERGED","created":"2020-06-02 19:09:06.000000000","updated":"2020-06-03 03:18:39.000000000","submitted":"2020-06-03 03:16:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732981-1591154217636-18d62acd","meta_rev_id":"9a731f200ac3b7c0aec967475543fc915661f1cc","_number":732981,"virtual_id_number":732981,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 03:16:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-06-03 02:11:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-06-03 02:11:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 02:11:04.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-06-03 03:16:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99f3c9fa446cec2cc1268ae102c80250841914dd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 19:09:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e415e7e695b6695d1dc141e8b6f75b33c12174ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 20:12:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b94c383d8d44d458fb83936c36da8aa : SUCCESS in 7m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a26a674074f426c99b1085b662cce80 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76b6a98897714b54adb56751f54550e0 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/092cf050e1c84a1989ca019a4b46a234 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3841dfea165b4d7e8d29c4ba4940553e : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a1ca8856eaf445a914ec86f63a2f651 : SUCCESS in 9m 57s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/33979cd2c85a4483aef9c626797b7ff2 : SUCCESS in 47m 41s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/21e5422bc52940329fcdcd4ed6359fbd : SUCCESS in 49m 00s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d08f68d2af69407f91bca817d5239e48 : SUCCESS in 1h 01m 24s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/a16a5c8fec564ea3ab970d234626da0e : SUCCESS in 35m 53s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/4808bef0f7c1458dbfc7f603d98be374 : SUCCESS in 45m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"831e4ca77fd1564761fa2e2ed7fa0d7909647e2e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-03 02:11:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks for the patch","accounts_in_message":[],"_revision_number":1},{"id":"1504fe59afd32839ff80f70380c9acbece2863a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 02:11:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"521481738062f051fa1101ccb2a549e446ede6cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 03:16:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/295eaec26c8c488a8223f81828bc0127 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7b9631cfa8541b1b4a066b677f04af0 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb05d8bbd3c84d84a054e9ddee8f41ee : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c743c9cf38c0414c90815d87f2965d0c : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c555249a2c4ba4ad3be9ca12b97ede : SUCCESS in 13m 18s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/49e0939545714e3d8bf42c3910d84799 : SUCCESS in 46m 36s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/c0a0128c571f4448b29c257f573e338f : SUCCESS in 50m 50s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/dd309529dfeb494183fdf8ed804f23fb : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f6c69122cd32a6fd21e00aa41ec5274af784d998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 03:16:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"faebdc6d948c45dba32d7a66a935577e0c7de9b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 03:18:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d755d41f85e413799fc2c049b5a8893 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90f120bebae1e9dcca88c1693da1eaaf2016f604","revisions":{"90f120bebae1e9dcca88c1693da1eaaf2016f604":{"kind":"REWORK","_number":1,"created":"2020-06-02 19:09:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/81/732981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/81/732981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/732981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/732981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/732981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/81/732981/1"}}},"commit":{"parents":[{"commit":"0e691ad6d6499bcfa7bbd03030415485396ff2d1","subject":"Merge \"Fix init container status - docker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0e691ad6d6499bcfa7bbd03030415485396ff2d1"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 19:08:59.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 19:08:59.000000000","tz":120},"subject":"Stop to use the __future__ module.","message":"Stop to use the __future__ module.\n\nThe __future__ module [1] was used in this context to ensure compatibility\nbetween python 2 and python 3.\n\nWe previously dropped the support of python 2.7 [2] and now we only support\npython 3 so we don\u0027t need to continue to use this module and the imports\nlisted below.\n\nImports commonly used and their related PEPs:\n- `division` is related to PEP 238 [3]\n- `print_function` is related to PEP 3105 [4]\n- `unicode_literals` is related to PEP 3112 [5]\n- `with_statement` is related to PEP 343 [6]\n- `absolute_import` is related to PEP 328 [7]\n\n[1] https://docs.python.org/3/library/__future__.html\n[2] https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n[3] https://www.python.org/dev/peps/pep-0238\n[4] https://www.python.org/dev/peps/pep-3105\n[5] https://www.python.org/dev/peps/pep-3112\n[6] https://www.python.org/dev/peps/pep-0343\n[7] https://www.python.org/dev/peps/pep-0328\n\nChange-Id: I564d6a8718feadd5f9d3cf12c73f66d13ec47d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/90f120bebae1e9dcca88c1693da1eaaf2016f604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/90f120bebae1e9dcca88c1693da1eaaf2016f604"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
