)]}'
{"id":"openstack%2Fzun~733993","triplet_id":"openstack%2Fzun~master~Id2e64b108a917a707bcc077fba187724f2fb48d9","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Id2e64b108a917a707bcc077fba187724f2fb48d9","subject":"[DO NOT MERGE] Test kuryr-libnetwork patch","status":"ABANDONED","created":"2020-06-07 20:53:58.000000000","updated":"2020-06-29 00:03:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f75c2bae5f39f5a61e2eacc56ce0c51535dcaa4c","_number":733993,"virtual_id_number":733993,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-18 14:29:56.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-16 15:09:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 15:09:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ed44b55b2611ed9a77741664bb01e6a6fa7b843","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-07 20:53:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4630c3b6bebc43766fab990e3d45474342faa1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-07 21:51:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ba314026edfe464783584a9542f1bc55 : SUCCESS in 7m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5d04368faf040d4ba34e3530371e1a4 : SUCCESS in 12m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dcfc08370a44ea9818a4baaca68ab4e : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a2d9e278264ed395d768c8a2879dbf : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46b627a814904703b8f8cf2920e6c49c : SUCCESS in 11m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1a3b24cfc8c450a88e26eca43696be0 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3ed06ec6658449da2af4e45dc0be499 : SUCCESS in 14m 42s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/3d91aa51022d4c06919a62f136369826 : SUCCESS in 42m 04s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/128400ae09724b8982eadceab0cebc80 : SUCCESS in 55m 57s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/8d115dec677843a6a78242a4417b9fac : SUCCESS in 54m 36s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/001023475fad4cb7a05246b631c29936 : SUCCESS in 42m 45s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/50135f5879d84bc292822b92b55f2170 : SUCCESS in 37m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bb849bbae735db20e1466bb9c935d17cfedb965","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-07 22:40:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"650377abe4d2c61d743b30c93dc7aa43fc014d2a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-08 05:11:14.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"89601778ca5717fcb01f8a966832e92c4fb21b7c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-08 05:11:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"007875e660409298a668f1960a04c137c6eb64a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 06:17:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cbeb3e518ba5485ca9e1bf92eb9d0164 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f26d7054ae8d487aa3730f77c0175db2 : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e932d4c21bd14aecb61dd86311096809 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31934ac7918346d89906e51bb47c0e97 : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/911b28dd2e1345f2b6891e4b78964a7c : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48c16e91ca8341e1b2502c7200c2558c : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07604bf7886a400798cf29317830c6af : SUCCESS in 11m 22s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b92c41fcad5b409ab409b4c735401a24 : SUCCESS in 48m 32s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/c3b9e1c64d39408f9ffd7a3368b660fd : SUCCESS in 49m 05s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/9e3c36098fd54d32839faa368392f7b2 : SUCCESS in 1h 04m 36s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/e21e08f2168547ffa682e0b35780b85e : SUCCESS in 35m 33s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/2f8d373effad4caca3aac93610057b4d : SUCCESS in 43m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77fa93b0ca6a4346b3f3e18b8ce39df9867e05d2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-16 14:23:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c40f065c357d510591595d28ca1203dc9d25f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 15:09:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bc9513ea1962495799ff8388aeea4f89 : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d873044431a443548d76faf5e5f4bced : SUCCESS in 14m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2aac23c638914839b1d3e1d9ae70bc46 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19b1d85dcb62471fbc0fc758cf7da409 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4d587dd246d4a5d88274477bb29fb08 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ecf154e84854f01baadd12c79104a90 : SUCCESS in 9m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b09c4ef2bdb420b9da6dd7f47f1fe2b : SUCCESS in 10m 17s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b6fadd389a3e44f2b4d8bb83ada49690 : FAILURE in 26m 40s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/656bd10c064241b7a11da5f0d94bd6b7 : FAILURE in 25m 04s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/c4325545fe924639bf068a8374314b1f : FAILURE in 38m 56s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/530ec88b28b642f8aa60685d5bfcd456 : FAILURE in 44m 13s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/28458aed9f1b41e391d687c1902cb462 : FAILURE in 29m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29dbffbc7ca5f22d4101f565c72bf6d9eb1bb578","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-18 14:29:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eafdc8971a0156b97787ea88c902ddf2aeb54754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 15:46:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6f446323d345440ba37cc2b965750d20 : SUCCESS in 8m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93a5167abd794a45b41c73cf1911b2ab : SUCCESS in 7m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c26115206b648728f63d483ad313fc0 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e077d2266b5041e1b60bbf132fbff130 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b70f3831a92541b486aa3b809043dce4 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d420eeb9119443cbb78ff81152c9243 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbb4d8adfe944e198b7c2d1b34dad9f2 : POST_FAILURE in 20m 23s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/13639225c4b54b718ed74bbf4884d60e : SUCCESS in 52m 19s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/2c41287abde54c258b10809647ef7be4 : SUCCESS in 45m 20s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/53b31a13374a46d6ad000982360dfef7 : SUCCESS in 1h 03m 44s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/f2375daf6d51446c9496e4955f05eebc : SUCCESS in 34m 19s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/45f0c02b0cc84900b0600ea1101b3d65 : SUCCESS in 45m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f75c2bae5f39f5a61e2eacc56ce0c51535dcaa4c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-29 00:03:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f51b01ddc6e577025b2ea7c127e45ca3a187da81","revisions":{"f51b01ddc6e577025b2ea7c127e45ca3a187da81":{"kind":"REWORK","_number":1,"created":"2020-06-07 20:53:58.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/93/733993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/93/733993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/93/733993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/93/733993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/93/733993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/93/733993/1"}}},"commit":{"parents":[{"commit":"90f120bebae1e9dcca88c1693da1eaaf2016f604","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/90f120bebae1e9dcca88c1693da1eaaf2016f604"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-07 20:53:25.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-07 20:53:25.000000000","tz":0},"subject":"[DO NOT MERGE] Test kuryr-libnetwork patch","message":"[DO NOT MERGE] Test kuryr-libnetwork patch\n\nDepends-On: https://review.opendev.org/#/c/733992/\nChange-Id: Id2e64b108a917a707bcc077fba187724f2fb48d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f51b01ddc6e577025b2ea7c127e45ca3a187da81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f51b01ddc6e577025b2ea7c127e45ca3a187da81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
