)]}'
{"id":"openstack%2Fzun~743671","triplet_id":"openstack%2Fzun~master~Ib30cdf698ae47b924aaf320954ddf7aee0d6aa2b","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ib30cdf698ae47b924aaf320954ddf7aee0d6aa2b","subject":"[WIP] Add job testing againest linux bridge","status":"NEW","created":"2020-07-29 03:01:29.000000000","updated":"2020-07-29 17:27:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e8b89168454e020efd413ecd71f277ca5d3d0cd2","_number":743671,"virtual_id_number":743671,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-29 17:27:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 17:27:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"442c50c3dea9dcee8fa67bc10305c81c867d33e8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-29 03:01:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ea6cc3234e9ba9f0beab7aee9be84c8ab67699e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 03:01:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/zun on branch master.  The error was:\n\n  Job zun-tempest-docker-iptables-linuxbridge not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python3-victoria-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - zun-tempest-py3-docker-sql\n          - zun-tempest-docker-sql-ipv6-only\n          - zun-tempest-multinode-docker-sql\n          - zun-fullstack:\n              voting: false\n          - zun-tempest-docker-sql-standalone:\n              voting: false\n          - zun-tempest-docker-iptables-hybrid:\n              voting: false\n          - zun-tempest-docker-iptables-linuxbridge:\n              voting: false\n      gate:\n        queue: zun\n        jobs:\n          - zun-tempest-py3-docker-sql\n          - zun-tempest-docker-sql-ipv6-only\n          - zun-tempest-multinode-docker-sql\n\n  in \"openstack/zun/.zuul.yaml@master\", line 162, column 3","accounts_in_message":[],"_revision_number":1},{"id":"9f857027d2b35a527e9d7fd91567b1d2e8e21981","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-29 03:13:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"72abe1d599901986a3bbda17ef4fbd308f70ec14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 04:34:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/328fdafd04e246ad8d4107262ff431b2 : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a819afb3ad7545468a64634950d8bd9e : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cf415a2559645c19ca77885fbc6bdf1 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ee676a784ff4b22b7ce3b44048fc4a4 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b5076d32b22431e8ff3ce4dd95f5614 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c977d1f6594c42bc4ece91cae12b0c : SUCCESS in 7m 44s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ee9bcc5174b64734b024a227695f0d5f : SUCCESS in 56m 57s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/dab701e79c7443b380672d4f5dd29b23 : SUCCESS in 43m 52s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/0f13c79e6c374788a396a83e7973f76c : SUCCESS in 1h 12m 03s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/415f7fe7645f4e1eaef778c5af3e8f24 : SUCCESS in 24m 42s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/db0441c1584f419aa064f38c660662a1 : SUCCESS in 45m 25s (non-voting)\n- zun-tempest-docker-iptables-hybrid https://zuul.opendev.org/t/openstack/build/4a449b31e9f847c8ad117ceeea387f70 : FAILURE in 50m 32s (non-voting)\n- zun-tempest-docker-linuxbridge https://zuul.opendev.org/t/openstack/build/d3ebd5d507144007a0ce6e1bd40a9409 : FAILURE in 36m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8e6bfab461ad2ea4df51af393d2292717052c9e8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-29 16:03:02.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e8b89168454e020efd413ecd71f277ca5d3d0cd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 17:27:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b40fb6411df49ef835b8fad93a53f6b : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3768ff158bb348c7ab2d21160db2c7db : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1913088f96ed41ff8060a1ee7d3c82db : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/600fafdff14649a28a3dcad88b6dec1b : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4eca4f4a12a4815b87fa3a6152f8fcc : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011f5ba1d8d84132905b638a3fa717b9 : SUCCESS in 5m 31s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7d8f256ebec94fefa4ad337f7a665eab : SUCCESS in 57m 34s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/631a92f46f70456e89275e8e76ba3605 : TIMED_OUT in 1h 13m 25s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/58e19b61cdd04918955a333f4e8a4d42 : SUCCESS in 1h 06m 19s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/c35065d08b3649a8b136d96864c1ec09 : SUCCESS in 32m 36s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/8445e8c052304661bef4949b1bfbb6a2 : SUCCESS in 53m 43s (non-voting)\n- zun-tempest-docker-iptables-hybrid https://zuul.opendev.org/t/openstack/build/f2c320cf4b134426bd668315ac274a9b : FAILURE in 1h 00m 54s (non-voting)\n- zun-tempest-docker-linuxbridge https://zuul.opendev.org/t/openstack/build/f213a1138475418e93d955ab398b060e : FAILURE in 44m 46s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9dec06ea8cb603a8a4d20fa4fde28d48c8ed3ba5","revisions":{"0f5530d5d5a15b7455cd965e2a1612e0a35be5ea":{"kind":"REWORK","_number":1,"created":"2020-07-29 03:01:29.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/71/743671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/71/743671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/743671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/743671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/743671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/71/743671/1"}}},"commit":{"parents":[{"commit":"799d67fc901dfc691e1a7281f0ed3295c3a715af","subject":"[WIP] Add iptables hybrid job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/799d67fc901dfc691e1a7281f0ed3295c3a715af"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-29 03:00:48.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-29 03:00:48.000000000","tz":0},"subject":"[WIP] Add job testing againest linux bridge","message":"[WIP] Add job testing againest linux bridge\n\nChange-Id: Ib30cdf698ae47b924aaf320954ddf7aee0d6aa2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0f5530d5d5a15b7455cd965e2a1612e0a35be5ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0f5530d5d5a15b7455cd965e2a1612e0a35be5ea"}]},"branch":"refs/heads/master"},"429b19be99493d4548e5e470589a2646b5e16cb4":{"kind":"REWORK","_number":2,"created":"2020-07-29 03:13:54.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/71/743671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/71/743671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/743671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/743671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/743671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/71/743671/2"}}},"commit":{"parents":[{"commit":"799d67fc901dfc691e1a7281f0ed3295c3a715af","subject":"[WIP] Add iptables hybrid job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/799d67fc901dfc691e1a7281f0ed3295c3a715af"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-29 03:00:48.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-29 03:13:46.000000000","tz":0},"subject":"[WIP] Add job testing againest linux bridge","message":"[WIP] Add job testing againest linux bridge\n\nChange-Id: Ib30cdf698ae47b924aaf320954ddf7aee0d6aa2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/429b19be99493d4548e5e470589a2646b5e16cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/429b19be99493d4548e5e470589a2646b5e16cb4"}]},"branch":"refs/heads/master"},"9dec06ea8cb603a8a4d20fa4fde28d48c8ed3ba5":{"kind":"NO_CHANGE","_number":3,"created":"2020-07-29 16:03:02.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/71/743671/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/71/743671/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/743671/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/743671/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/743671/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/71/743671/3"}}},"commit":{"parents":[{"commit":"86d1afb636e326857666f13adc7aba7bc98079cd","subject":"Add iptables hybrid job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/86d1afb636e326857666f13adc7aba7bc98079cd"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-29 03:00:48.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-29 16:03:02.000000000","tz":0},"subject":"[WIP] Add job testing againest linux bridge","message":"[WIP] Add job testing againest linux bridge\n\nChange-Id: Ib30cdf698ae47b924aaf320954ddf7aee0d6aa2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9dec06ea8cb603a8a4d20fa4fde28d48c8ed3ba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9dec06ea8cb603a8a4d20fa4fde28d48c8ed3ba5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
