)]}'
{"id":"openstack%2Fzun~744101","triplet_id":"openstack%2Fzun~master~I22a700b5f90fa43f602fda42bd4f9bd8f64b1d72","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I22a700b5f90fa43f602fda42bd4f9bd8f64b1d72","subject":"Fix mock name for capsule ut","status":"MERGED","created":"2020-07-31 07:57:09.000000000","updated":"2020-07-31 15:53:47.000000000","submitted":"2020-07-31 15:52:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744101-1596210749643-b464fa81","meta_rev_id":"a28ea914d5f44967c50ac1430136f843bfa82ccd","_number":744101,"virtual_id_number":744101,"owner":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-31 15:52:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-07-31 14:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-07-31 14:38:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 14:38:56.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-07-31 15:52:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bd4f6f75bb2ab767ccc5acc310714ce3b345ef2","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2020-07-31 07:57:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe18ddbdd858976a0cace2e954a8d89768fe0b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 09:14:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/822dded902cc418a8f2c5ea91a0da886 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4af816441fd84cd98162261b481bdd01 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e9ed1db6284345a552f1dba9910996 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23bba62ee5e044d79766cb9fdc90425d : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a0662b274f549ea9400f42fecd7dc85 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c3253a14e274d38bdd9ba215ef6cdef : SUCCESS in 6m 18s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/1a29f3b0bd6f46638364eb88baee2982 : SUCCESS in 49m 01s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/002c0404dd924b098350e96bedbd7c18 : SUCCESS in 1h 07m 51s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/51ea3505fbf945859787cdedfb086aa5 : SUCCESS in 58m 58s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/f17cb6dec0e34317bd982673037896e7 : SUCCESS in 32m 43s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/e3c7b63ea63f4455b8a6b02a70d4b8e4 : SUCCESS in 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10caeb34013e19f2f971bbe9a74db2eae3e0aeb2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-31 14:38:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"322e41851e41baf41c5d47e1ba202928920bcf17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 14:39:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04d4bcc1f343df91f2e3d727d1e662834be06e54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 15:52:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6ee1e9014254971a123a7f746f32e22 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e194b4cad14f9da79179987a2ea712 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f2590526b4d4d4dacefabefa12d85eb : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cf98dd8eac34b9db4ed68ca9be7533c : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da04b1fb43624cd3bfaf1235e7a94074 : SUCCESS in 6m 07s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2889dc5bab384816b38c2fe2ed696719 : SUCCESS in 52m 26s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/fefabd871d834673926c0262e2935d4c : SUCCESS in 48m 30s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/80074c20cf8b48e69befc588c68a7538 : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ef3fca0fc8bf07d98fcf8fea48abfcbdda9567cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 15:52:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a28ea914d5f44967c50ac1430136f843bfa82ccd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 15:53:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f308b149ea441ca9050a9c2166633d : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f6fd3f5f7551091aac44d4b6941b550b4dab429","revisions":{"9f6fd3f5f7551091aac44d4b6941b550b4dab429":{"kind":"REWORK","_number":1,"created":"2020-07-31 07:57:09.000000000","uploader":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"ref":"refs/changes/01/744101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/01/744101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/744101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/744101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/744101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/01/744101/1"}}},"commit":{"parents":[{"commit":"c2a72d89001e07873f3accc06b35fe6d12a894db","subject":"Merge \"Rename network_api to network_driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c2a72d89001e07873f3accc06b35fe6d12a894db"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-07-31 07:50:39.000000000","tz":480},"committer":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2020-07-31 07:50:39.000000000","tz":480},"subject":"Fix mock name for capsule ut","message":"Fix mock name for capsule ut\n\nChange-Id: I22a700b5f90fa43f602fda42bd4f9bd8f64b1d72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9f6fd3f5f7551091aac44d4b6941b550b4dab429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9f6fd3f5f7551091aac44d4b6941b550b4dab429"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
