)]}'
{"id":"openstack%2Fzun~744576","triplet_id":"openstack%2Fzun~master~I6e9b14db8fd9efa3e07e2365d4c503aea2484884","project":"openstack/zun","branch":"master","attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-08-31 01:42:01.000000000","reason":"A robot voted negatively on a label"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-31 01:42:01.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6e9b14db8fd9efa3e07e2365d4c503aea2484884","subject":"[DO NOT MERGE] Test CNI network driver","status":"NEW","created":"2020-08-03 23:11:04.000000000","updated":"2026-08-31 01:42:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b1897dd4446c75a8e47bc37a73d53a2c187b3bfa","_number":744576,"virtual_id_number":744576,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-31 01:42:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-20 16:45:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47c605924b76f038899efadff6377812fbdb87f6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-03 23:11:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26a128aa1ce227cf52d154e29d49965aafbe0a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 00:46:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f91f7ed946b04bad89d78d08f29855ef : FAILURE in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b01c60c53b924a4789789982a9c57375 : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e8778f4d18a44699a6950b09fed5701 : FAILURE in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1ded9166178433a9181b67324d5f160 : FAILURE in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbfa5ec6cc7b4340b4aa326de4b2de7e : FAILURE in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/297573041bc342c29306a412173577f4 : SUCCESS in 6m 38s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/bbde3ab20a9b4a648382f1db4392eb1f : FAILURE in 1h 00m 11s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/76fcfa5054134fea961101a1c63e2e5e : FAILURE in 57m 09s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/7c1a8e376cb34d729515cdb42d395081 : FAILURE in 1h 19m 56s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/3861213233374f6a8aa1e95ab2e85c70 : FAILURE in 46m 43s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/9ccbf3ab80d84033bf746c57a9ec8251 : FAILURE in 38m 28s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/network/cni_network.py","accounts_in_message":[],"_revision_number":1},{"id":"c3107a489366e99b37d843bb61d07bbd1684dd66","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-04 02:35:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9ff223ad7bd010be7f088487854b86348a0a02c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 03:58:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8fc85f259f64c5fae30f7fbe19b8097 : FAILURE in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/392ce50920ff4c4e8ce5e3d87b369ca4 : FAILURE in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81ffa112acda464aad3099dcaef51842 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c19c4d5be3f143318ef8353098ba5003 : FAILURE in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb66ead131ea47d28436874e75ea7c5a : FAILURE in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f8fad0e7c0c4940a422bfb89ccce39f : SUCCESS in 5m 48s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/fd0aaeba5f4c448f848e5b05e17303d9 : FAILURE in 1h 08m 25s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/31b7ecf557014a00887c088b85581028 : FAILURE in 58m 28s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/cb26219d25fb48f2b4dc6268e7529e3d : FAILURE in 1h 12m 05s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/4316c938986d4b84b7e31520485dabe2 : FAILURE in 40m 13s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/b8c529e2915046edb8eca8d14cfba680 : FAILURE in 54m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d4a9ed6696945e35db74515a95226027ebbd57e9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-09 03:26:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3070bf94b7a748b4eeb1b206a82a1d29399f910a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-09 04:35:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9e55437d8e143c08e46f5dfac7374fa : FAILURE in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b46dfb65a80c453d87075206693b789b : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e8b936123e4bce9b719ecd7e3be989 : FAILURE in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c65491fbf46e466f978164f45558883d : FAILURE in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4af95e17bb549b69f1f2f10d4ea5cb4 : FAILURE in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f84c917e6bb4e2b812d8e790824c2be : SUCCESS in 6m 23s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/4a1c04660a5444e5ad278d784b7e1e95 : FAILURE in 49m 21s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/a8cc41683f694faea9b145dcec53aa56 : FAILURE in 52m 27s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/218d10f1076f42b78318938a31556de4 : FAILURE in 1h 03m 29s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/7d765b2bf3924809860a86066825b5a2 : FAILURE in 44m 44s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/cc26e44604bd43688b670ef3092f76e2 : FAILURE in 46m 49s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/network/neutron.py","accounts_in_message":[],"_revision_number":3},{"id":"dfbb109fd300e0d3e20624d4e2573388d0813f45","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-09 05:51:40.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"a4f0240d85957fdc770599b49d475d4b752b9784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-09 06:59:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b65269a714d143a2bfad9846eedf17d4 : FAILURE in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5eb1966259e4772bdad7e39670440d3 : FAILURE in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9a84740307f4bca850f5a869d68d635 : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6e7c9daf7aa429fbb76246432419fcb : FAILURE in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/372b48f8843d4131b7aae8f9a9eb34c1 : FAILURE in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1d347fc147a42c091190af6f232f862 : SUCCESS in 5m 28s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/fb9bc634c2e94ceda77dd2a766e82e70 : FAILURE in 50m 12s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/69f076ac9ba54fe08a7c3260b8e9c2b9 : FAILURE in 53m 24s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/5dd33b47009246108a221e14f99c3de9 : FAILURE in 1h 06m 26s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/bd53597953454a739e44ccbfdfab5eb2 : FAILURE in 38m 01s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/62a2c4ba0e404a33b93972b431b69e79 : FAILURE in 48m 05s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/network/neutron.py","accounts_in_message":[],"_revision_number":4},{"id":"2aad99f6106aa50346b20e5448a3af87eb36cde9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-07 03:37:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2fda6274cd1ed02f698d3bc0d7b001befbe1f919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 05:00:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7fb119298584c59a17528dbed321f64 : FAILURE in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/513defbd9cf3497d91724339462d2b81 : FAILURE in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900a1ed1f0e4422ab55120c3df81911d : FAILURE in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89b083662dd542beb93a4e675f517e30 : FAILURE in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4def9c2bc864038aa2443a031bb8c22 : FAILURE in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da1d1c21e8a8487f9b0cef80d2e1fdbf : SUCCESS in 13m 07s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d1f7a43552604885ba8a07aceeb78db8 : FAILURE in 1h 07m 40s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e2b50d8b23bc422092c6562f57dbb627 : FAILURE in 1h 05m 29s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d187bc8dc2524b75b3e35778c76539cb : FAILURE in 1h 21m 13s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/d1a7e39b52f44b23898d60e649cc769e : FAILURE in 51m 21s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/546fe309ddd9428bb5890f9807fa1517 : FAILURE in 1h 00m 36s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/network/cni_network.py","accounts_in_message":[],"_revision_number":5},{"id":"e31b20912bd2aeedef3e40890df44de3869bdf9f","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-07 05:16:52.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"a1c7a68c2b3d3f089092d71269eefaa95eb91c0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 06:37:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a36504dab6b1486e96e05e5f1532dc6d : FAILURE in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60f0f76f95b44b4ca6c32465bce3c605 : FAILURE in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfe8178303f5425fb616843e6bad801b : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1144be32894488e8ee46f46eb0938dd : FAILURE in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91e8fb00c0d541378985396aac33e728 : FAILURE in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d445d1bc304b6c87180c235682e171 : SUCCESS in 5m 18s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ce56445c1cac4ab6af9c41ede9239699 : FAILURE in 1h 00m 30s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/b236382e93554da99323795eb6d9fe98 : FAILURE in 59m 19s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2f368a001abb4301be457437d095eb26 : FAILURE in 1h 13m 33s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/22aa94837841432182631d2b87178b69 : FAILURE in 55m 19s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/29652ce8847b4cfba43fac2ab32c1070 : FAILURE in 58m 47s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/network/cni_network.py","accounts_in_message":[],"_revision_number":6},{"id":"7b884353b9920eb46a3bdfa1ffe8c8a0c39767b7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-08 04:16:21.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"a52540d7739d958adc2eb932a3cc22d183218ee6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-08 04:46:38.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"00437b11e8c70bac584413e49fc0722b58f0427b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 06:07:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d0266a75ab148bc8bca60f4bea72c3c : FAILURE in 4m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2d0748e80f540149090d26d26ad9d8f : FAILURE in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac996e6330c842db8a8248a1306e8293 : FAILURE in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b787de54087460caca5b611553949cd : FAILURE in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c0b683e693e49b18affdcc0b876d057 : FAILURE in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a175d1e1b1fc4e20ad66ad3fdaef333d : SUCCESS in 4m 48s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/cc31f70606224ee29861b1d19ff74f78 : FAILURE in 59m 38s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/9eab24d20eee4b1d99b4558d241ef1d7 : FAILURE in 55m 10s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d440de43072f4f178949dea4158fb65b : FAILURE in 1h 13m 59s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/bc8b90b67ae04a74ba7bc8bcb67db5e8 : FAILURE in 44m 19s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/0612f0d66c714095a8ca311c72dd09d9 : FAILURE in 53m 23s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/network/cni_network.py","accounts_in_message":[],"_revision_number":8},{"id":"fc13983883b8ab85f6cb3cd001fb1ff5ef1b0ac4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-08 14:32:26.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"57d4f6584a206f52861bc6712d84d19aebef0bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 15:49:36.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40c849d2d68240c2aae3d2358195075d : FAILURE in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1a37fe72c47446991cd01a3e80b26aa : FAILURE in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00c128043ffd46d196ce43820a541110 : FAILURE in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f99c01befa054696ad298a3d3dc3b275 : FAILURE in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25dc33ebb8064c54ae11aa90481f7632 : FAILURE in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee4fe8c177364c27add2c10b9090de2f : SUCCESS in 8m 17s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/0fd5a5f0af9649188d768b04fa624e81 : FAILURE in 57m 21s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/260b9b1228a048f1966103302dad334d : FAILURE in 52m 24s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/5da05f18512047b385d657a948a6b694 : FAILURE in 1h 06m 56s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/c47db14a449b4628bef8b028c428fae4 : FAILURE in 44m 54s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/b911b8ba557a41abb3417dbb155573d8 : FAILURE in 59m 17s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/network/cni_network.py","accounts_in_message":[],"_revision_number":8},{"id":"cbad2c8a13a66c1a8ccd1b9c6c0b0ffc07c1c384","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-12 23:38:45.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"17119e06ca0b39ac37754426623432603f02953f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 00:51:17.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e825d1289d134f86b377a386a1548a8e : RETRY_LIMIT in 3m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5ec1d45a05e43738323139406f66405 : RETRY_LIMIT in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d13f6c64a8244b9bb8255027a21ca57 : FAILURE in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aafe6b6be084d40a0d0bbe25555fadb : FAILURE in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8468c735a0740bf8ae83d4d7fb6d7df : RETRY_LIMIT in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f1de5a8c6a41dbb8d860290f080e47 : SUCCESS in 13m 18s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/db7bda2317394194a331866ea077df75 : FAILURE in 58m 13s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/55be4d43067d4f02a3de5395a1e7b573 : FAILURE in 55m 07s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/4311cddb4d0d4a0291acabf49a3bc8bb : FAILURE in 1h 06m 54s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/1f9892c2b8a24883a9105e7cd8773f0b : FAILURE in 42m 10s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/0c1418f575094210ad37491d2e763301 : FAILURE in 50m 54s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/tests/unit/api/controllers/test_base.py\n  Comments left for invalid file zun/api/controllers/v1/capsules.py\n  Comments left for invalid file zun/api/controllers/v1/containers.py","accounts_in_message":[],"_revision_number":9},{"id":"0fd7a1bf771290b1c923bf143b499c981637b4a2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-13 03:06:44.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"d9aab51fb600c22f501483ce9383e00f955249ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 05:11:35.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b1e02abb30445f4a8aaed98878a82ff : RETRY_LIMIT in 3m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57cb1021246c4f41bf3fc17f20428eaf : RETRY_LIMIT in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61757b71d74e47b2beaa6cf9516cb9bf : FAILURE in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b09c85f27494e4caac464215e88024e : FAILURE in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a77cceaec85458a8f72a350a1c223d0 : RETRY_LIMIT in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b56a93633fc343aeb07870621a0a2b86 : SUCCESS in 16m 21s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/418d9c7aebd0455796448b8b2b2dbef3 : FAILURE in 55m 47s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/31a89f8dcbec4f78a9fa248d83efb79b : FAILURE in 56m 25s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/3e19a7633d3c4231bee85fcc8a1f3731 : FAILURE in 1h 07m 15s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/f53cc69846544ecfbc0817c44fab8265 : TIMED_OUT in 2h 02m 52s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/e0f35eaf0fc0437cb869dc7b56924691 : FAILURE in 52m 40s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/tests/unit/api/controllers/test_base.py\n  Comments left for invalid file zun/api/controllers/v1/capsules.py\n  Comments left for invalid file zun/api/controllers/v1/containers.py","accounts_in_message":[],"_revision_number":10},{"id":"e20e4816a856d7582a41124605ada446a073267c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-20 04:17:30.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"3622d55eef57f9ca0a3dcd9bab8ae82e5edf527f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 05:34:27.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6f10a1045da464a9d1061e9e33492b2 : FAILURE in 5m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9712c3a4b1de421c913d64e976f95c9f : FAILURE in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0cc9590e90b4b399d0fa4e826204087 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adda44f2e38040729ce4dd53aed74542 : FAILURE in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fec0430d722346d4869f1c56e75cc4c5 : FAILURE in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090783f692834719b26c29c67dcdfc29 : SUCCESS in 15m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/f4daf0845432448dbceb56f3a0d87411 : FAILURE in 56m 18s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/6d01c0b6fed448eb8e35cbe39123c16b : FAILURE in 46m 23s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/50c392ef578944d491e721a070019cc2 : FAILURE in 1h 10m 24s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/9a943849195c4580858426e24959a76d : FAILURE in 44m 04s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/26fc2d8c76e04001861243e3c55b6a84 : FAILURE in 53m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6ba43439679a89139af3110bebc5d19d28e3bf7d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-20 15:49:36.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"5e4eab62aafb004b5aaf25229e56b4551ab7c972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 16:45:22.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17a420c0359d436991a63b421f3f386b : FAILURE in 4m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8208d5acd8894b7aa26c1631ae70ad41 : FAILURE in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/139bf6eacce7418e97308364819dd195 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b61053689e5449248127da20b0bb7364 : FAILURE in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cce59c19be8b4d289eecfb3debdb2b54 : FAILURE in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da3f81543afd478e843ede91f2de623c : SUCCESS in 11m 29s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/84d8042a8c694b7a970679c3e593a249 : FAILURE in 45m 24s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/7e94c2d74b5242268b9b0841ad228cfc : FAILURE in 45m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/3440276e26c94460b53281fd54e64b1d : FAILURE in 54m 49s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/20083f35acdd4efc88e3aa34d3f3f236 : FAILURE in 33m 36s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/ab7ac85a861d48f8ae5daa8be575a7b5 : FAILURE in 43m 46s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d4921f352f2e3e101c3168ac8fa6d692014ce13a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-12-29 10:57:43.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"35611501ed6308624d82b459955187fb30e58f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 12:01:55.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d315b5c4444d4580961ad58dcd9ae5c8 : FAILURE in 4m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d32f2b6b1fc45c3b18e28567a5e4604 : FAILURE in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a6676c0f05747c1a886708bbd9449e1 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11c19573298940e8b57dfce6ac5f2d72 : FAILURE in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2e6867e20cd43e29712f8e1081a6695 : FAILURE in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d33a725f79ab49e6b678e067a46689c9 : FAILURE in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7fdf597070e48438920e0bf61d63975 : SUCCESS in 5m 35s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7a82263e5e4246bea9fa81e8ef7c3da3 : FAILURE in 1h 00m 18s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/509f348c34384441b064d82773fa06a0 : FAILURE in 50m 33s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/184352e850e84475b82dd7e5ebb55d91 : FAILURE in 1h 03m 06s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/30f08455fbe54690a4d42fa77e29961c : FAILURE in 44m 26s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/0fe540537a9a4bf681b5bb195c15dbc6 : FAILURE in 42m 36s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3c571ae979e55b285fc2c28005a6969135028153","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 00:10:50.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"9382c5da5e5225c2e1e3826d5e01680ca2d59eae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 00:10:54.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"cd39dd8817a5843b02d9c159bf7a2d387f5da18a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 00:28:00.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"1c4f7a85e91a60a0d36786952cd66653afbd2769","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 00:28:15.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"b1897dd4446c75a8e47bc37a73d53a2c187b3bfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 01:42:01.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf7bed35068f44de9e4efbd5e9726a0a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/484b628261da4e439b9eab583f7d1a04 : FAILURE in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78e491964ff6480fbb1173be701db62c : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73bc99559ff543dfbbd6b58736c95333 : FAILURE in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c58f44c986e545f1b40815dc78cc251e : FAILURE in 4m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/af4bed7889384bb79038c880500cd679 : FAILURE in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46d15e51078841208e54636e023b85b5 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95a7a59824b94c21a8112e5fbf89aece : SUCCESS in 4m 30s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/c5c73045b2a645b3a59e9f8f8e65c1b1 : FAILURE in 50m 22s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/603e17fb972c4b7593c3cf47f29aa0c7 : TIMED_OUT in 1h 12m 43s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d0efea8046be47ba8e0f54ae8e9a0e4f : FAILURE in 38m 09s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/635ef3c204f9474db341b07c61784115 : FAILURE in 18m 11s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/8f67861bea5b4214a912dd20aab2b7ae : FAILURE in 45m 58s (non-voting)","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"13fc8633d1ff81089dfd391879c68a60cb7a2bcc","revisions":{"d02502ca42ed1a138a08233df79c61f9023c19bd":{"kind":"REWORK","_number":1,"created":"2020-08-03 23:11:04.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/1"}}},"commit":{"parents":[{"commit":"41eafec7d6434edede207daf761c87b28774a648","subject":"[WIP] Switch docker to CNI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/41eafec7d6434edede207daf761c87b28774a648"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d02502ca42ed1a138a08233df79c61f9023c19bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d02502ca42ed1a138a08233df79c61f9023c19bd"}]},"branch":"refs/heads/master"},"01b71c89450ecfff144ee3b503068863b3c6be4a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-04 02:35:41.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/2"}}},"commit":{"parents":[{"commit":"6dcbcf316f3b9563ac331bfbecbc104dbca263a5","subject":"Introduce CNI network driver for docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6dcbcf316f3b9563ac331bfbecbc104dbca263a5"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-04 02:35:41.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/01b71c89450ecfff144ee3b503068863b3c6be4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/01b71c89450ecfff144ee3b503068863b3c6be4a"}]},"branch":"refs/heads/master"},"211e84e478d12770c37c31e80b47c2130f60dd77":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-09 03:26:18.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/3"}}},"commit":{"parents":[{"commit":"ab95996119435a3fba40bc35656c0081580dbb64","subject":"Introduce CNI network driver for docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ab95996119435a3fba40bc35656c0081580dbb64"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-09 03:26:18.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/211e84e478d12770c37c31e80b47c2130f60dd77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/211e84e478d12770c37c31e80b47c2130f60dd77"}]},"branch":"refs/heads/master"},"96baca28cc61f5421ff91dd0a3fd398f321412df":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-09 05:51:40.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/4"}}},"commit":{"parents":[{"commit":"9aab368d6ec58c7eec49dcc5aeb73ec3a451e20f","subject":"Introduce CNI network driver for docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9aab368d6ec58c7eec49dcc5aeb73ec3a451e20f"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-09 05:51:40.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/96baca28cc61f5421ff91dd0a3fd398f321412df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/96baca28cc61f5421ff91dd0a3fd398f321412df"}]},"branch":"refs/heads/master"},"75e6d6aa6b625fc85a34f2b7dae4e7a8073f9ce0":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-07 03:37:13.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/5"}}},"commit":{"parents":[{"commit":"f9f4dbf5bd1e4a0aa4f52b7ebef116f171b8ad38","subject":"Introduce CNI network driver for docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f9f4dbf5bd1e4a0aa4f52b7ebef116f171b8ad38"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-07 03:37:13.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/75e6d6aa6b625fc85a34f2b7dae4e7a8073f9ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/75e6d6aa6b625fc85a34f2b7dae4e7a8073f9ce0"}]},"branch":"refs/heads/master"},"41132009283f82ac8b46b1cc3c92d11f7ed2b5ec":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-09-07 05:16:52.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/6"}}},"commit":{"parents":[{"commit":"7b8fc1b106cf4b17c3a1410e61ff5c4f23398d1a","subject":"Introduce CNI network driver for docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7b8fc1b106cf4b17c3a1410e61ff5c4f23398d1a"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-07 05:16:52.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/41132009283f82ac8b46b1cc3c92d11f7ed2b5ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/41132009283f82ac8b46b1cc3c92d11f7ed2b5ec"}]},"branch":"refs/heads/master"},"fdcb65c7cf7f633102ff4142723ad3f4a08cee30":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-08 04:16:21.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/7"}}},"commit":{"parents":[{"commit":"b80d4702b4e3d690f1d3640cc9c95f49112725ff","subject":"Introduce CNI network driver for docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b80d4702b4e3d690f1d3640cc9c95f49112725ff"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-08 04:16:21.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/fdcb65c7cf7f633102ff4142723ad3f4a08cee30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/fdcb65c7cf7f633102ff4142723ad3f4a08cee30"}]},"branch":"refs/heads/master"},"81c09416e8ad20c0766d21472c0494b0ffb89a52":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-09-08 04:46:38.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/8"}}},"commit":{"parents":[{"commit":"7034da52b2734c5be05e2492779e9de452448b44","subject":"Introduce CNI network driver for docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7034da52b2734c5be05e2492779e9de452448b44"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-08 04:46:38.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/81c09416e8ad20c0766d21472c0494b0ffb89a52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/81c09416e8ad20c0766d21472c0494b0ffb89a52"}]},"branch":"refs/heads/master"},"6fe12e469c3736cd468a1078d9b019cce7c362bb":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-09-12 23:38:45.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/9"}}},"commit":{"parents":[{"commit":"8c8e0a55351e34b2cf768fadea00d828699d735b","subject":"Introduce CNI network driver for docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8c8e0a55351e34b2cf768fadea00d828699d735b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-12 23:38:45.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6fe12e469c3736cd468a1078d9b019cce7c362bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6fe12e469c3736cd468a1078d9b019cce7c362bb"}]},"branch":"refs/heads/master"},"0bbf7364f607212520867906154148bcaf03aa4c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-09-13 03:06:44.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/10"}}},"commit":{"parents":[{"commit":"a5f3e3ffc19f7b5ed61d15bd726165591d95727f","subject":"Introduce CNI network driver for docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a5f3e3ffc19f7b5ed61d15bd726165591d95727f"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-13 03:06:44.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0bbf7364f607212520867906154148bcaf03aa4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0bbf7364f607212520867906154148bcaf03aa4c"}]},"branch":"refs/heads/master"},"245dd889aad2329ca46ff0856dfae14b32bf916d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-09-20 15:49:36.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/11"}}},"commit":{"parents":[{"commit":"25ee5972ec0c05da7632988e2d86bc8d343c32c3","subject":"Introduce CNI network driver for docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/25ee5972ec0c05da7632988e2d86bc8d343c32c3"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-20 15:49:36.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/245dd889aad2329ca46ff0856dfae14b32bf916d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/245dd889aad2329ca46ff0856dfae14b32bf916d"}]},"branch":"refs/heads/master"},"784f61830f6ef453f5bc8b67988a62b63745a46e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-12-29 10:57:43.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/76/744576/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/12"}}},"commit":{"parents":[{"commit":"1bce9e5890ed7f76c273e60a672ea853ca252875","subject":"Fix lower constaint job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1bce9e5890ed7f76c273e60a672ea853ca252875"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2020-12-29 10:57:43.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/784f61830f6ef453f5bc8b67988a62b63745a46e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/784f61830f6ef453f5bc8b67988a62b63745a46e"}]},"branch":"refs/heads/master","description":"Rebase"},"306dec52ec80d62536c401729292eca87a1f4d0c":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2026-08-31 00:10:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/744576/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/13"}}},"commit":{"parents":[{"commit":"1bce9e5890ed7f76c273e60a672ea853ca252875","subject":"Fix lower constaint job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1bce9e5890ed7f76c273e60a672ea853ca252875"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-31 00:10:50.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/306dec52ec80d62536c401729292eca87a1f4d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/306dec52ec80d62536c401729292eca87a1f4d0c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"728c57f44fa344c8c497d1921d32ef27fe3546ca":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-08-31 00:10:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/744576/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/14"}}},"commit":{"parents":[{"commit":"09834b7977b5ec1e6baa2471e13e4e657b379fee","subject":"Merge \"bindep: Remove stale markers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/09834b7977b5ec1e6baa2471e13e4e657b379fee"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-31 00:10:54.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/728c57f44fa344c8c497d1921d32ef27fe3546ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/728c57f44fa344c8c497d1921d32ef27fe3546ca"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"1bce9e5890ed7f76c273e60a672ea853ca252875","ours":"306dec52ec80d62536c401729292eca87a1f4d0c","theirs":"09834b7977b5ec1e6baa2471e13e4e657b379fee","merge_strategy":"recursive","contains_conflicts":false}},"94dfd6a375464c3f592da8e5c385849809a31329":{"kind":"REWORK","_number":15,"created":"2026-08-31 00:28:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/744576/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/15"}}},"commit":{"parents":[{"commit":"09834b7977b5ec1e6baa2471e13e4e657b379fee","subject":"Merge \"bindep: Remove stale markers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/09834b7977b5ec1e6baa2471e13e4e657b379fee"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-31 00:25:36.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/94dfd6a375464c3f592da8e5c385849809a31329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/94dfd6a375464c3f592da8e5c385849809a31329"}]},"branch":"refs/heads/master"},"13fc8633d1ff81089dfd391879c68a60cb7a2bcc":{"kind":"REWORK","_number":16,"created":"2026-08-31 00:28:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/744576/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/76/744576/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/76/744576/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/76/744576/16"}}},"commit":{"parents":[{"commit":"09834b7977b5ec1e6baa2471e13e4e657b379fee","subject":"Merge \"bindep: Remove stale markers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/09834b7977b5ec1e6baa2471e13e4e657b379fee"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 23:10:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-31 00:28:15.000000000","tz":0},"subject":"[DO NOT MERGE] Test CNI network driver","message":"[DO NOT MERGE] Test CNI network driver\n\nChange-Id: I6e9b14db8fd9efa3e07e2365d4c503aea2484884\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/13fc8633d1ff81089dfd391879c68a60cb7a2bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/13fc8633d1ff81089dfd391879c68a60cb7a2bcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
