)]}'
{"id":"openstack%2Fzun~746381","triplet_id":"openstack%2Fzun~master~Ibe99a2bd89c3e66ffe5ddb6502e734b6dddbf872","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ibe99a2bd89c3e66ffe5ddb6502e734b6dddbf872","subject":"[DO NOT MERGE] Test OVN with Zun","status":"NEW","created":"2020-08-15 23:44:18.000000000","updated":"2026-09-09 13:03:56.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8794a1d2d23f91b61cedca15c26c310b6f9afd08","_number":746381,"virtual_id_number":746381,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 13:03:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 15:11:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7a32af4b15d123e4b91a7a05c60e52cd01bdf2a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-15 23:44:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14f394e997f70254308980dfac0a2f491f1c3ba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 00:42:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9d8274058d84d6db29df1271ccb0bf9 : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69ffb7ebb895429aaef739f794aa59f1 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7babcae72b948b48b03af29b35bf98b : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7fd0b27f567410398a5091d6bef219d : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bb83a988abc4bc4836411b0e517044c : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf96a7e51904ec8966e3697fc58e696 : SUCCESS in 7m 17s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/44fb7d89102a47119ed0fc593e5ffc51 : SUCCESS in 47m 07s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/89d9ddd4468744b38d8caecd0ebf59fe : SUCCESS in 48m 56s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/fd6a7f018c8e47d39ba79579ca9b6d63 : SUCCESS in 56m 02s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/9ea19c43e06b4574a298d67adecab7a1 : SUCCESS in 32m 09s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/7c8ea30db6f54417b1f51ab67a953fc6 : SUCCESS in 40m 31s (non-voting)\n- zun-ovn https://zuul.opendev.org/t/openstack/build/7f729962e8d845288080e0e1898d0516 : SUCCESS in 49m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ded25a9aaebba2b5032908654c8ebcdd58cfe9ee","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-24 13:54:19.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7662cb8b29f2de96330d9e7e5ca2f9b94cc4fa43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 15:11:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae2207a1006745c98879ecd6bf70b1f7 : SUCCESS in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14465689e74a4ad4bfcc15fc00675342 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec4d897cb2d545d0aa3cede6537f51c0 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a022e8e140145faa7508901081ca1e7 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/395d945ad06c422995e9ba1312dde0af : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3614daabc584889b6c72e758c6db102 : SUCCESS in 10m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/28035f826b53453b9accd1938341d8fb : SUCCESS in 46m 31s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/42e12cc3d23848ae8b362a7cf91c9ce3 : SUCCESS in 1h 02m 51s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/19d97253de1049e59c9f31d52c135551 : SUCCESS in 1h 07m 05s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/584327e6fa4f4f12b5e6881804800f93 : SUCCESS in 38m 02s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/8cb16005885e4021ba8ff13f2e07ed34 : SUCCESS in 1h 02m 24s (non-voting)\n- zun-ovn https://zuul.opendev.org/t/openstack/build/7282b9453d204f4fa8c7f53efac7d8c2 : SUCCESS in 58m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b288460aec5f142bac6daef30855a927b297232b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 01:51:36.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"61d5b8357fe7b158532c668570fd66f61836ccad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 01:51:41.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"6c50eac9b207a42e51ff4a874fc639f530e74df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 02:41:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85609ff46bfb4a46925cd9f8a8d9ce22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce492b7f3d56400a80eb120b222e22ff : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95ad453794724710b14fda2910caebd8 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6fa6847402945c7a2b498e3f07304d1 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7724ee5338bf41538549f4fb2820db16 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9bdb2bbcea6b40cebc1c18bdf72ffc90 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d36f33ca254847bf8cadd13e1ed770 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20e420620a7b4ce8829a9ddf2279f8a5 : SUCCESS in 6m 19s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/3fb63f5a10cf489f9b134443ea704a16 : SUCCESS in 35m 50s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/bd6a3e31b1624f58a2598b7ce68a5f9b : SUCCESS in 43m 43s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/e34c2f147701400a98f7e82f6a312dae : SUCCESS in 47m 47s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/4aeba4aaff2f420a8cabe1d8a8c109ef : FAILURE in 34m 27s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/2b495c7e0801420cb987caf6e64f1073 : SUCCESS in 40m 15s (non-voting)\n- zun-ovn https://zuul.opendev.org/t/openstack/build/656d13befb624683a0a7d383e41877df : SUCCESS in 38m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3e696af6e634f1888191c95f6505e3ecf6d2ee2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-09 12:05:58.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"8794a1d2d23f91b61cedca15c26c310b6f9afd08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 13:03:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2854cf0386e444dc99500478f26128bf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e92aa2a70804a07b5720fe6e27dee57 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/716ef80625984e898c0dacb0c789652e : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d63dad575a345fbb96c268443bd14b5 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2c6f71001554fcfac96614b3c0dd34f : SUCCESS in 6m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8814f7b94aed4c5a936373acc5a5b22f : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1ca43a01694f0b9ad357efa79cf87b : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c6adfaba8ea44a49833af63f61fb306 : SUCCESS in 7m 12s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/547a1c5349b24e2abd9b9fbcd1d5c675 : SUCCESS in 25m 18s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/3666fdc2ab15491c8c122c1d28577eb7 : SUCCESS in 31m 45s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/968ab0aca4d445af97608f882b9fe021 : SUCCESS in 39m 01s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/72442e6aab9849d8903e5f96f20ebb04 : FAILURE in 32m 07s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/432afc9d51aa4b11a9f0c92c5d801fd8 : SUCCESS in 30m 28s (non-voting)\n- zun-ovn https://zuul.opendev.org/t/openstack/build/ad20bc8910d14c648e3ddf324432b950 : FAILURE in 51m 41s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"111d22eddb8c89146a7fb47081793993ed5f71e3","revisions":{"a78253751cd0faca87236433d180945789c27807":{"kind":"REWORK","_number":1,"created":"2020-08-15 23:44:18.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/81/746381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/81/746381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/81/746381/1"}}},"commit":{"parents":[{"commit":"05d55ec9bdafb7ab06923c428906b0051a79c383","subject":"Refactor kuryr libnetwork code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/05d55ec9bdafb7ab06923c428906b0051a79c383"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-15 23:43:01.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-15 23:44:17.000000000","tz":0},"subject":"[DO NOT MERGE] Test OVN with Zun","message":"[DO NOT MERGE] Test OVN with Zun\n\nChange-Id: Ibe99a2bd89c3e66ffe5ddb6502e734b6dddbf872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a78253751cd0faca87236433d180945789c27807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a78253751cd0faca87236433d180945789c27807"}]},"branch":"refs/heads/master"},"3808fe153305454add9ad45f655e0139163bfba7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-24 13:54:19.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/81/746381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/81/746381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/81/746381/2"}}},"commit":{"parents":[{"commit":"05d55ec9bdafb7ab06923c428906b0051a79c383","subject":"Refactor kuryr libnetwork code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/05d55ec9bdafb7ab06923c428906b0051a79c383"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-15 23:43:01.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-24 13:54:15.000000000","tz":0},"subject":"[DO NOT MERGE] Test OVN with Zun","message":"[DO NOT MERGE] Test OVN with Zun\n\nDepends-On: https://review.opendev.org/#/c/746435/\nChange-Id: Ibe99a2bd89c3e66ffe5ddb6502e734b6dddbf872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3808fe153305454add9ad45f655e0139163bfba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3808fe153305454add9ad45f655e0139163bfba7"}]},"branch":"refs/heads/master"},"4259e2c1d4add98fc83166e32c8e9d582cfd110d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-01 01:51:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/746381/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/81/746381/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/81/746381/3"}}},"commit":{"parents":[{"commit":"05d55ec9bdafb7ab06923c428906b0051a79c383","subject":"Refactor kuryr libnetwork code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/05d55ec9bdafb7ab06923c428906b0051a79c383"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-15 23:43:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 01:51:36.000000000","tz":0},"subject":"[DO NOT MERGE] Test OVN with Zun","message":"[DO NOT MERGE] Test OVN with Zun\n\nDepends-On: https://review.opendev.org/#/c/746435/\nChange-Id: Ibe99a2bd89c3e66ffe5ddb6502e734b6dddbf872\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4259e2c1d4add98fc83166e32c8e9d582cfd110d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4259e2c1d4add98fc83166e32c8e9d582cfd110d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"70fa1818db75aef9083c84717fd6b2854a8e5627":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-01 01:51:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/746381/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/81/746381/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/81/746381/4"}}},"commit":{"parents":[{"commit":"09834b7977b5ec1e6baa2471e13e4e657b379fee","subject":"Merge \"bindep: Remove stale markers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/09834b7977b5ec1e6baa2471e13e4e657b379fee"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-15 23:43:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 01:51:41.000000000","tz":0},"subject":"[DO NOT MERGE] Test OVN with Zun","message":"[DO NOT MERGE] Test OVN with Zun\n\nDepends-On: https://review.opendev.org/#/c/746435/\nChange-Id: Ibe99a2bd89c3e66ffe5ddb6502e734b6dddbf872\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/70fa1818db75aef9083c84717fd6b2854a8e5627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/70fa1818db75aef9083c84717fd6b2854a8e5627"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"05d55ec9bdafb7ab06923c428906b0051a79c383","ours":"4259e2c1d4add98fc83166e32c8e9d582cfd110d","theirs":"09834b7977b5ec1e6baa2471e13e4e657b379fee","merge_strategy":"recursive","contains_conflicts":false}},"111d22eddb8c89146a7fb47081793993ed5f71e3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-09 12:05:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/746381/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/81/746381/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/81/746381/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/81/746381/5"}}},"commit":{"parents":[{"commit":"5da5e0a4d96ac220fa59a6039463177170b9213c","subject":"Merge \"Update Docker remote API version and remove tempest_exclude_regex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5da5e0a4d96ac220fa59a6039463177170b9213c"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-15 23:43:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-09 12:05:58.000000000","tz":0},"subject":"[DO NOT MERGE] Test OVN with Zun","message":"[DO NOT MERGE] Test OVN with Zun\n\nDepends-On: https://review.opendev.org/#/c/746435/\nChange-Id: Ibe99a2bd89c3e66ffe5ddb6502e734b6dddbf872\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/111d22eddb8c89146a7fb47081793993ed5f71e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/111d22eddb8c89146a7fb47081793993ed5f71e3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"09834b7977b5ec1e6baa2471e13e4e657b379fee","ours":"70fa1818db75aef9083c84717fd6b2854a8e5627","theirs":"5da5e0a4d96ac220fa59a6039463177170b9213c","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
