)]}'
{"id":"openstack%2Fzun~748337","triplet_id":"openstack%2Fzun~master~I0dfe1c5311b16c63b9c766411acdb75126cd0561","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I0dfe1c5311b16c63b9c766411acdb75126cd0561","subject":"Remove mox3 from lower-constraints.txt","status":"MERGED","created":"2020-08-27 01:37:25.000000000","updated":"2020-08-28 03:57:41.000000000","submitted":"2020-08-28 03:56:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748337-1598586990593-46488824","meta_rev_id":"623659364667340f0087dbbead5ecdbd0459e5ae","_number":748337,"virtual_id_number":748337,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 03:56:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-08-27 02:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-08-27 14:42:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":1,"date":"2020-08-27 05:00:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-08-27 02:07:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 02:07:23.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-08-27 05:00:58.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-08-27 14:42:03.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-08-28 03:56:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a0594b10f7e6c1ff24f2e388da9bfb7a9c1e9ea","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-08-27 01:37:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c155ec34d2530ed42ae8a136410faa1de70510c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-27 02:07:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a1858853eacf66b4a2d8ab9f84a40232884752e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:55:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6df73d83c4d9474e8ba9ad23aea184c6 : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4ddc7ee4c2047afb132f092ce6b5a58 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b17aedef8e7e4a03bd1fe47ba055a81e : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b462407a8944c0ac9d2599f068545f : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0efe6ddbc4744158bdb0e5d71b629668 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6690bf22fa3d4da9b4f785a0ff69a32b : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa7e2d7be3614ed5a0670cf8f3949358 : SUCCESS in 5m 42s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a9dca0c898714b04ae3813f3b09cea4a : SUCCESS in 53m 37s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/17a24ed1f4fc4d9aa27efac91efd3127 : SUCCESS in 50m 45s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/1db79308746d43c49832c6b6aa595c53 : SUCCESS in 1h 07m 42s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/c5e738861f6844ef8f9961e0df2f1f59 : SUCCESS in 41m 22s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/b0a65da41a5f445aa27f2b974a1c37e0 : SUCCESS in 41m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7345d8a96e87812a3be1265c60c6a2f05e1cca2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:55:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b9631c25f7a139458a6f73a8f2a69d3c38c9742","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 03:08:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f4e872826f2c487dbf1a3fb44678ef0e : SUCCESS in 2m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c317402a0fd447699080f7fa0d7d318 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7403a72f06644df6a040114aa1972c3e : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6efbf593815f4b91902146706adde94a : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4a2ecf22e434c3f8022d1e963a8eb25 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b0cc61540084c2b9741ef8a9745c83e : SUCCESS in 5m 13s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a4cc0f3fb03b4bd79f3f92e8bafeb1cc : FAILURE in 10m 15s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/fbc32c3eda9b4918993ea781e2c2b754 : FAILURE in 8m 49s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/1953916a59104305aae09a6a8e785907 : FAILURE in 11m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e2b7db4eb67a339421e4aa4c62922a1aad3c2686","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-08-27 05:00:58.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc467c6ea85d856f5e6a92be69051b1bbd09bac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 06:25:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c7bf5b5f0fec42d7a9a1ad5180a4b3b3 : SUCCESS in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dd5f0e401584157b03d269ff31457b8 : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5aab9f8eb09d4abd9fb35bf580594703 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad3e9c5e45664aad8520fb511cfd4fc4 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/309a78dbda074892b3bfe42c892f1be1 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/131afc6511664018945439cecd5c532d : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37bb7e8b7ad404fb207c5f3eec730d8 : SUCCESS in 6m 51s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/efd827a285684e7282f021e104e2bbbf : SUCCESS in 43m 18s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/ccd8316996ee49b9bcdb1bb0f928cdaa : FAILURE in 9m 02s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/fa8728509a7c455fabcc3c4a4a591672 : SUCCESS in 1h 23m 16s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/9adbeea62b15462caaa885eddf5764b9 : SUCCESS in 37m 29s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/847b85d5a8324853b73c5d29523d4649 : SUCCESS in 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b76d1d09a7e3b7dc69b62bbb9918c804d1ed1d3c","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-08-27 14:42:03.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f6d38e7e8f149c19145ca00514ce83876fe8db13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 16:04:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/acc81d68668e4962a5489851e2e0a05c : SUCCESS in 4m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8fb4bc51afa414896ee2e415b4bca15 : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/553d11a8a15a4812820a9c185c385fb7 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fefe3c9b8854fabbde0513fd732d661 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7baa772d505a4e66ad6490d2da6b3b7c : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d584203125e44d44bff54cb6f42a4c57 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/758d5f77ccdd4188bac3d33cbe394fc3 : SUCCESS in 6m 36s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7e1ce86674174feea431cf8e7d731f97 : SUCCESS in 53m 02s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/0310421cff93495e96ea371c8e741f10 : SUCCESS in 52m 55s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/cd00cb88c18e4970880d5088143ce264 : FAILURE in 1h 05m 25s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/a4f5cdd718ee4aa0b117cf6a8eeaa072 : SUCCESS in 41m 29s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/e7e0814601a14c1db4d3c89367481675 : SUCCESS in 47m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45000eb2526b7f5da2a8c8fd0082629a9e8e3353","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-08-28 01:46:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"891943a51f818c1d70be8c35b4f103346ef2a1c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 02:53:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e004aa6ab5a44904ac25663b9997edb8 : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50d2082418fa4953bc4b45f8734412ff : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a5de2b8a4a647d89fb58c75d3c5ee4f : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b609d42b5a4640a39ce890b5f8e4be76 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29e647506a1344069604b4757245d544 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3302cf4f26c84663ba6602122ad6d77a : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20910f0467d340df9a8897c1b5e77272 : SUCCESS in 5m 54s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/372938d0c09f495a82a0f9ec8c98dc7b : SUCCESS in 49m 52s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/9b49c73f701f41b281a913d40287059c : SUCCESS in 52m 10s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/00eaf8429d2e4b0b918dbc354e21870c : SUCCESS in 1h 01m 26s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/2102cad32ad14a4486c818af2365e4f5 : SUCCESS in 39m 39s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/733c0b0ab8a644caa8f58d980f55e327 : SUCCESS in 48m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8578a67991cfd83458361ccc862f2f5a23778086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 02:53:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d54640e1941c1ae94cb3e555a32558bfab78b2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 03:56:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6325a0bc09a48d2bef808ced59d7ca9 : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0e6d6af03dc499e9e2bea302001a7c2 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/987e2ab6c810431290a68561f809b840 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adb810236e834ea9a6b3751965ae9e55 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/830b78f9e43c4614a59abad6179aa8cc : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc47e7c3aa84b019f807f674bcdefa3 : SUCCESS in 6m 50s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/3ec5158e100a469ab547e665fbde2ca4 : SUCCESS in 46m 29s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/5c6b6224894b4bf083f6bbd8bb692a82 : SUCCESS in 47m 15s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/e523da037a744fab8478f688e001dad3 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2606a221dfdeb64ea876a26e08d244c4741d10f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 03:56:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"623659364667340f0087dbbead5ecdbd0459e5ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 03:57:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edbbda080ac14cc99ec2a3fe89039547 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40416bd23fced8f8a7ca17d52b2e5ddff5d8a63a","revisions":{"40416bd23fced8f8a7ca17d52b2e5ddff5d8a63a":{"kind":"REWORK","_number":1,"created":"2020-08-27 01:37:25.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/37/748337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/37/748337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/748337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/748337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/748337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/37/748337/1"}}},"commit":{"parents":[{"commit":"a8e97c118c016ee2320f05fb442d5850585f8ccd","subject":"Merge \"Add kuryr-libnetwork as a required dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a8e97c118c016ee2320f05fb442d5850585f8ccd"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-08-27 01:35:23.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-08-27 01:37:22.000000000","tz":480},"subject":"Remove mox3 from lower-constraints.txt","message":"Remove mox3 from lower-constraints.txt\n\nThe mox3 repo is being retired.\n\nChange-Id: I0dfe1c5311b16c63b9c766411acdb75126cd0561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/40416bd23fced8f8a7ca17d52b2e5ddff5d8a63a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/40416bd23fced8f8a7ca17d52b2e5ddff5d8a63a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
