)]}'
{"id":"openstack%2Fzun~748947","triplet_id":"openstack%2Fzun~stable%2Fussuri~I5a244892d46b69be1394eb89a86a4c16fcd1439d","project":"openstack/zun","branch":"stable/ussuri","hashtags":[],"change_id":"I5a244892d46b69be1394eb89a86a4c16fcd1439d","subject":"api-ref: fix wrong naming in /execute","status":"MERGED","created":"2020-08-31 04:52:05.000000000","updated":"2020-09-10 01:48:02.000000000","submitted":"2020-09-10 01:46:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748947-1599702398090-210ac784","meta_rev_id":"0d5bcee32e1c6eadf0636064a82d14a5c31e7356","_number":748947,"virtual_id_number":748947,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 01:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-09-10 00:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-09-10 00:50:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 01:46:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5fd59e331c670008f95e306164c2710578024c3","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-31 04:52:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6e67121e59cdc13d98ca8e533affe2aefd996985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 04:58:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e85fc425cab14abc9765cdfadcc23e41 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d72d0467d04dd5825ea856aa75bd17 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0b51f610a29f43ba5a6422c3a4aed4eab0d2d968","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-10 00:50:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87bebff22a6c15c7f5f9d6973751ab2ac1be01ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 00:50:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df5b80547d4e44d3f81e033b21f242297bcaedbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:46:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d5e17fc1ab4e1ba4b414aeb47d5024 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf766858dadc4195848fecd78db81748 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0b5aef563b22b192d690b75888571de16e586e14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:46:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0d5bcee32e1c6eadf0636064a82d14a5c31e7356","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:48:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd144b11f1ef44cc914c76799b342ba8 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d41a395f0b8a02c7c5970f4b54d77c22d548986","revisions":{"5d41a395f0b8a02c7c5970f4b54d77c22d548986":{"kind":"REWORK","_number":1,"created":"2020-08-31 04:52:05.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/47/748947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/47/748947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/47/748947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/47/748947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/47/748947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/47/748947/1"}}},"commit":{"parents":[{"commit":"d5511fd828832886a37fd2d3cb133942b81c2a8e","subject":"Dep\u0027s should be restricted by upper-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d5511fd828832886a37fd2d3cb133942b81c2a8e"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-24 04:44:31.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-31 04:52:05.000000000","tz":0},"subject":"api-ref: fix wrong naming in /execute","message":"api-ref: fix wrong naming in /execute\n\nChange-Id: I5a244892d46b69be1394eb89a86a4c16fcd1439d\n(cherry picked from commit 0a3a763e1bdc53e7a4b82c7cf06b177c38d84f7c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5d41a395f0b8a02c7c5970f4b54d77c22d548986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5d41a395f0b8a02c7c5970f4b54d77c22d548986"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
