)]}'
{"id":"openstack%2Fzun~751955","triplet_id":"openstack%2Fzun~master~I46a8c67a4d5ff97d03db097c17864061befe501c","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I46a8c67a4d5ff97d03db097c17864061befe501c","subject":"Several fixes to unblock the gate","status":"MERGED","created":"2020-09-15 02:25:22.000000000","updated":"2020-09-20 03:13:36.000000000","submitted":"2020-09-20 03:12:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751955-1600571547259-515a9e8b","meta_rev_id":"dd277930c0ebdc9a1e06cfd0bb5551d1a4ed66e8","_number":751955,"virtual_id_number":751955,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-20 03:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-09-20 02:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-09-20 02:12:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-20 03:12:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2fb3648cc3f1f23a4f4f5044b028209f2bb86121","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-15 02:25:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de36cce0d50056a02ab388a3c56f532d4ffec28d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 02:50:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f327fce3a3df46ab88a6aba50182eee1 : FAILURE in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed22338b1c8e4f699e934ab1a3f39513 : FAILURE in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c43fcc0bbe94840b305f675acf17d87 : FAILURE in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/505ce190210548cfa8cc331028fcb169 : FAILURE in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74f63b12017a4a9c8b412f49a346e5a3 : FAILURE in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02464bebbc624342b53c69f8858576ef : SUCCESS in 15m 50s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/f3d38efb26024a3e9da9099ff77bd62b : FAILURE in 14m 31s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/20b2208abeb94f77ae22b2c2cdfed433 : FAILURE in 15m 27s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b7572772c2e945f7afcb75a20dfd5463 : FAILURE in 17m 50s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/1e89bb7e0b3a4939818f205fa225b4e7 : FAILURE in 13m 58s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/7cd1b718908644cd96b0b801cb8b1355 : FAILURE in 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"893b65a2df9927a0d5c48f5008f84c16ebddf290","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-15 13:11:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36f728d63991939de56ba604002260abc4b78ec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 13:54:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5315041863354c89aaca64c905f78dde : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e918eb2a330404fa252827496684f91 : POST_FAILURE in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7492c0e7014c78aabef017b8245eab : FAILURE in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e72812846a8547408efd7e12e59605a8 : FAILURE in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d93c96d9638c4dfc800e2e262abccba2 : POST_FAILURE in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f61e85a9e04f64995c708ee7cdb19b : SUCCESS in 17m 13s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/3bbe1e5d82ac42d6a38421d88719ce14 : FAILURE in 13m 56s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/0d8e7ada5961479ab8c4251f5badc9f9 : POST_FAILURE in 19m 50s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/7d95d4e6c2084775981c5f1833cfff71 : FAILURE in 22m 20s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/87ac1c096b144f44b5a4e5312e034cd8 : SUCCESS in 37m 20s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/ef10007d7e1f4f45b0518b4fd1264692 : FAILURE in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5436bcb687c5cafc6e2b7ee296b8a522e61f92d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-15 16:11:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a198f5e0693a022bf736adbab725949ed296a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 17:42:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65f1385d80a8454db1dccd19e1b2f576 : SUCCESS in 7m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/556910c94c8d43b884e367f0a876aeac : FAILURE in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70e7335965b4ab6ae1bbbed126e388f : FAILURE in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81818d5c880f4b878517f4af1fdf0f77 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4b8378c7168488dad29a03467435128 : SUCCESS in 18m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213492a120d74593b2239194c2d74a54 : SUCCESS in 23m 26s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ca30742e6fd443d1b36841e62fa47cba : FAILURE in 11m 15s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/a89c74d8cfef45c5857a57b56786895c : FAILURE in 11m 19s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/5a4da6c40c8b43109bc264ab3aa76ec6 : SUCCESS in 1h 27m 23s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/857240407ba44f2c947b395d036d4f64 : FAILURE in 12m 36s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/870b6bdb6f5e43779956ce95f9190596 : SUCCESS in 56m 36s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/api/controllers/v1/containers.py\n  Comments left for invalid file zun/api/controllers/v1/capsules.py\n  Comments left for invalid file zun/tests/unit/api/controllers/test_base.py","accounts_in_message":[],"_revision_number":1},{"id":"c1b896bbcfe9f008d67646e6fc0c08f5e940c36c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-16 01:31:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f8c8a723707efb1d0b8afd7060a061022a530b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 02:30:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d03fba270c794b43af66173d522a0cd9 : SUCCESS in 8m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cc568365e4049618cea887ae8be21dc : FAILURE in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f46dd633ec496395fa0cb8df3b7c5a : FAILURE in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c91dc9be68140a7a4a493be59c4446e : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35a0615e4d0b484aa2f3eeacbcc221ef : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d4e52cfc4c8486d9d702e36d6774255 : SUCCESS in 18m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/76db606836e84cd9b0332746cf84212c : SUCCESS in 44m 05s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/826db0a548a84cbd9df640b9c20eb650 : SUCCESS in 48m 18s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/235435b6622b4187ad3856154c8af27c : SUCCESS in 57m 39s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/e1cd01389d894e72a9822df71e96018a : SUCCESS in 34m 44s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/40f0caf2842c49149ff090370c6445f2 : SUCCESS in 48m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file zun/tests/unit/api/controllers/test_base.py\n  Comments left for invalid file zun/api/controllers/v1/capsules.py\n  Comments left for invalid file zun/api/controllers/v1/containers.py","accounts_in_message":[],"_revision_number":1},{"id":"7c05cc09b90a41af1d0c8042a74f3cf6b917098b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-19 19:27:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"69cd732de850c7238fd4c9e3f0bbafa2f4157805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 20:39:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ce58e09eb023489d99233e95f7bed774 : FAILURE in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5c6019db66a405497da9d80d2b9301f : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/479af420cb654a178a1a73e270382e28 : FAILURE in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5b78a24e3874912b1e8883a736c6207 : SUCCESS in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea2c6a1b4b27442fa26a9d388cfe0cd8 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dc22cdedd49461c8c32a43ef7f7641d : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0b3807fcdb04c13ab25629bae26c51f : SUCCESS in 25m 51s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/e9748d8192044019aebebb9684857e67 : SUCCESS in 47m 27s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/059bdca5e24c4f139d26772b986162d2 : SUCCESS in 36m 36s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/9b8ffdd665484d49bbd410952a05beb5 : SUCCESS in 1h 10m 16s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/b32d431042534957ad538f6b3decb20f : SUCCESS in 41m 48s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/ab59648ea3774b289f28cdfcbf9db2c8 : SUCCESS in 51m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"198c381f0f3d27a1f2782d4f26866a2c2719c03a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-19 22:03:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0c10c23a8d97a745f3896a024fd20703c5c25048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 23:06:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/81ae41a8e9a8418595217a33e1580310 : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05840778535f4bb38f1591414bcfb1bb : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f1ba8487c834531a58233f79f6c7534 : FAILURE in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc3fcae5dccf465d80a81094e81402ca : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23d8251754254ad293f587f5aed25460 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c1e75538cac4c6f9bee38241e7ca813 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b8e6e4f11c49fa9d44ac5233b00bc2 : SUCCESS in 14m 52s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b16b5d0010f34fc79afdbf9cc54a1a7e : SUCCESS in 46m 32s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/18f9915ab8c740adb5bf5af360610dcf : SUCCESS in 46m 54s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/42910b0f0a804fed9cbc242df34bf4d0 : SUCCESS in 1h 01m 17s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/14eb3f6252754294ab45f5342348049b : SUCCESS in 37m 34s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/8bace2752c9f4a8f8dc8a7e7865daa97 : SUCCESS in 42m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c758b1d9c4c47d8a3094b2b78addb05515c7a096","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-19 23:33:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"224759e3a0f75caa0bd16a35797ca4730e9d5842","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-20 00:23:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dbb09ab3631a4f612a1a7bce0ca7a574cf5f1244","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-20 01:12:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7cb844889488b9c795b09fe86348ec5c40add8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 02:09:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2cd0567f46aa498586a88d72af897d98 : SUCCESS in 6m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f6620cb0bf814e99bbde4937561a90a8 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/700525de158e48c8a2a15d699a88f984 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b849458ceb834e57854cebab9f4f92b5 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b13f10504f54d138614e2184d092a14 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/504c752899a64ca6af4d63d601268c74 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad03e7f3b8264e8f9d148f229c804625 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1be50269d6f349638ac260b6c1735f33 : SUCCESS in 13m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50f8aa5f4b474629bad6d9b9f35c8620 : SUCCESS in 3m 15s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/3735030d4c864b7da5e7d94e7b5f1f56 : SUCCESS in 38m 35s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/9b950e7766254fe0905d5d4612e60cbb : SUCCESS in 36m 25s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/760b2c1fc6ad4efab61358b9d7c27d85 : SUCCESS in 55m 08s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/ffcaf0aea9f7406d9374277b6ac20a5a : SUCCESS in 23m 46s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/8cada66243b74cc39bb546a5b65ed7a4 : SUCCESS in 36m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"98a26f5e73c1dd1516dd1ed735bcf84999754966","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-20 02:12:23.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"8ea77a6a8f6bdf9670478449a975af5298d3c03a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 02:12:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b663869d1267129624279aa1ca6f6ff9bbd73d2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 03:12:26.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7a1be9294ad34333aa4f5f107d5bd07f : SUCCESS in 5m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f0276cb54d634018bfc9b420c1442c93 : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5fb2b0e29d1417486a956e70da6b7f7 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ca883fdc604bf4bc106ae0d6b62d43 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75dd6549d4164769b8fefd1dc05a73d2 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/202f8ac0eacd4c40ad82e05e8d9509c1 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d74209aa486a41edaecc8b8773e689bc : SUCCESS in 14m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf2ad86b37884e8a98cbaed8a5651da8 : SUCCESS in 4m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2411301a998442ed808c48b7734c180a : SUCCESS in 43m 51s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/a3651113c7c5408890364ab8f0400628 : SUCCESS in 43m 03s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/6ad90c7cb8cf40afb30d3ce6d3741d67 : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":6},{"id":"00c0d6158035b2dc6e6996864601d2f1b5bc32d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 03:12:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"dd277930c0ebdc9a1e06cfd0bb5551d1a4ed66e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 03:13:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0c5e979096c8489d81050d6a8855f190 : SUCCESS in 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d736d0095e664a3d9ccb9689e3e4d55c : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5e3841f248d47689ae3da4e3d6cc43c : SUCCESS in 40s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b7b8fcd51eb09656c39ecac2bb0ea8df7a003e4e","revisions":{"e5996bba0c656d66690c3b1bca6a6733d3133662":{"kind":"REWORK","_number":1,"created":"2020-09-15 02:25:22.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/55/751955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/55/751955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/55/751955/1"}}},"commit":{"parents":[{"commit":"1f7419b2ba49ecdf964f2ec524ec839543d412ea","subject":"Merge \"api-ref: fix wrong naming in /execute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1f7419b2ba49ecdf964f2ec524ec839543d412ea"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-15 02:24:56.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-15 02:24:56.000000000","tz":0},"subject":"Make test-setup.sh compatible with mysql8","message":"Make test-setup.sh compatible with mysql8\n\nStarting from mysql version 8 it\u0027s not possible to create a user\nimplictly when using GRANT.\n\nThis patch makes the behavior compatible with that.\n\nChange-Id: I46a8c67a4d5ff97d03db097c17864061befe501c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e5996bba0c656d66690c3b1bca6a6733d3133662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e5996bba0c656d66690c3b1bca6a6733d3133662"}]},"branch":"refs/heads/master"},"e8d97a3431ec514135c5e7b9c657d1855b1b1ef2":{"kind":"REWORK","_number":2,"created":"2020-09-19 19:27:07.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/55/751955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/55/751955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/55/751955/2"}}},"commit":{"parents":[{"commit":"1f7419b2ba49ecdf964f2ec524ec839543d412ea","subject":"Merge \"api-ref: fix wrong naming in /execute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1f7419b2ba49ecdf964f2ec524ec839543d412ea"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-15 02:24:56.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-19 19:25:21.000000000","tz":0},"subject":"Several fixes to unblock the gate","message":"Several fixes to unblock the gate\n\n* Make test-setup.sh compatible with mysql8. Starting from mysql\n  version 8 it\u0027s not possible to create a user implictly when using GRANT.\n* Fix lower constraint for python3.8. See:\n  https://bugs.launchpad.net/nova/+bug/1886298\n* Fix pep8 failure\n\nChange-Id: I46a8c67a4d5ff97d03db097c17864061befe501c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e8d97a3431ec514135c5e7b9c657d1855b1b1ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e8d97a3431ec514135c5e7b9c657d1855b1b1ef2"}]},"branch":"refs/heads/master"},"992d62c2ccb162a9e1be293837e89910933adfa5":{"kind":"REWORK","_number":3,"created":"2020-09-19 22:03:59.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/55/751955/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/55/751955/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/55/751955/3"}}},"commit":{"parents":[{"commit":"1f7419b2ba49ecdf964f2ec524ec839543d412ea","subject":"Merge \"api-ref: fix wrong naming in /execute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1f7419b2ba49ecdf964f2ec524ec839543d412ea"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-15 02:24:56.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-19 22:03:54.000000000","tz":0},"subject":"Several fixes to unblock the gate","message":"Several fixes to unblock the gate\n\n* Make test-setup.sh compatible with mysql8. Starting from mysql\n  version 8 it\u0027s not possible to create a user implictly when using GRANT.\n* Fix lower constraint for python3.8. See:\n  https://bugs.launchpad.net/nova/+bug/1886298\n* Fix pep8 failure\n\nChange-Id: I46a8c67a4d5ff97d03db097c17864061befe501c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/992d62c2ccb162a9e1be293837e89910933adfa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/992d62c2ccb162a9e1be293837e89910933adfa5"}]},"branch":"refs/heads/master"},"f63d02e155d860f6b5d095557f855a662bc6a70b":{"kind":"REWORK","_number":4,"created":"2020-09-19 23:33:07.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/55/751955/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/55/751955/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/55/751955/4"}}},"commit":{"parents":[{"commit":"1f7419b2ba49ecdf964f2ec524ec839543d412ea","subject":"Merge \"api-ref: fix wrong naming in /execute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1f7419b2ba49ecdf964f2ec524ec839543d412ea"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-15 02:24:56.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-19 23:33:02.000000000","tz":0},"subject":"Several fixes to unblock the gate","message":"Several fixes to unblock the gate\n\n* Make test-setup.sh compatible with mysql8. Starting from mysql\n  version 8 it\u0027s not possible to create a user implictly when using GRANT.\n* Fix lower constraint for python3.8. See:\n  https://bugs.launchpad.net/nova/+bug/1886298\n* Fix pep8 failure\n\nChange-Id: I46a8c67a4d5ff97d03db097c17864061befe501c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f63d02e155d860f6b5d095557f855a662bc6a70b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f63d02e155d860f6b5d095557f855a662bc6a70b"}]},"branch":"refs/heads/master"},"f2375ff9231c770f08a459a1ad03b9df4625ab32":{"kind":"REWORK","_number":5,"created":"2020-09-20 00:23:41.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/55/751955/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/55/751955/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/55/751955/5"}}},"commit":{"parents":[{"commit":"1f7419b2ba49ecdf964f2ec524ec839543d412ea","subject":"Merge \"api-ref: fix wrong naming in /execute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1f7419b2ba49ecdf964f2ec524ec839543d412ea"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-15 02:24:56.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-20 00:23:33.000000000","tz":0},"subject":"Several fixes to unblock the gate","message":"Several fixes to unblock the gate\n\n* Make test-setup.sh compatible with mysql8. Starting from mysql\n  version 8 it\u0027s not possible to create a user implictly when using GRANT.\n* Fix lower constraint for python3.8. See:\n  https://bugs.launchpad.net/nova/+bug/1886298\n* Fix pep8 failure\n\nChange-Id: I46a8c67a4d5ff97d03db097c17864061befe501c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f2375ff9231c770f08a459a1ad03b9df4625ab32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f2375ff9231c770f08a459a1ad03b9df4625ab32"}]},"branch":"refs/heads/master"},"b7b8fcd51eb09656c39ecac2bb0ea8df7a003e4e":{"kind":"REWORK","_number":6,"created":"2020-09-20 01:12:49.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/55/751955/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/55/751955/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/55/751955/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/55/751955/6"}}},"commit":{"parents":[{"commit":"1f7419b2ba49ecdf964f2ec524ec839543d412ea","subject":"Merge \"api-ref: fix wrong naming in /execute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1f7419b2ba49ecdf964f2ec524ec839543d412ea"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-15 02:24:56.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-20 01:12:44.000000000","tz":0},"subject":"Several fixes to unblock the gate","message":"Several fixes to unblock the gate\n\n* Make test-setup.sh compatible with mysql8. Starting from mysql\n  version 8 it\u0027s not possible to create a user implictly when using GRANT.\n* Fix lower constraint for python3.8. See:\n  https://bugs.launchpad.net/nova/+bug/1886298\n* Fix pep8 failure\n\nChange-Id: I46a8c67a4d5ff97d03db097c17864061befe501c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b7b8fcd51eb09656c39ecac2bb0ea8df7a003e4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b7b8fcd51eb09656c39ecac2bb0ea8df7a003e4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
