)]}'
{"id":"openstack%2Fzun~753192","triplet_id":"openstack%2Fzun~master~I40e5f01f6df9f318faafe50ee1bf9c91b727859a","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I40e5f01f6df9f318faafe50ee1bf9c91b727859a","subject":"Add release note for CNI plugin of docker.","status":"MERGED","created":"2020-09-21 23:44:22.000000000","updated":"2020-09-22 16:42:23.000000000","submitted":"2020-09-22 16:39:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753192-1600792745369-365cc730","meta_rev_id":"ad14049a4eab8b6a01ee68fe4b1ceb7af50336eb","_number":753192,"virtual_id_number":753192,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-22 16:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-09-22 03:00:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-09-22 03:00:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 16:39:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d6b97f55b1511c2783c895180cddb094919cdff","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-21 23:44:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e000b174b25b3e542915745e768b6ff4045cd31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 00:58:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f092a442e1664aae88d273ad871d605c : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b787398de74ba0b32bc966fbfce5b9 : SUCCESS in 13m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8acc48b56c2f40a081ac88fda9dbd12e : SUCCESS in 4m 55s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ff18d42bd6c84cf0afd655647963c867 : SUCCESS in 51m 07s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/a70be9b2cf1241e2bf39f542f9bfa3d3 : SUCCESS in 59m 35s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/26d7f5e48b7748b39a02a16b2ada0ea4 : SUCCESS in 1h 05m 36s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/a60d854d9fa4413e8011bc8ab7eb85ab : SUCCESS in 40m 02s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/438df6a1b046452eb2c27fd4142e63cc : SUCCESS in 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8670910a3e0a6bd46b4e2e1922bc3d4046d6e21","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-22 03:00:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9c7c1a01fef5a0bbe10f114f1aabbea93e2cd4ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 03:00:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e17cb04ae465109cf16be8a15d233f6afcdb268d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 04:15:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acefb48ca05948f9b5680ebe44f3509d : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13e48d122c9f4ecfb69b99d6e846586a : SUCCESS in 15m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c75aca122d4e41fd935b4e49c5fc5806 : SUCCESS in 3m 50s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/68ad5f1bac8a4edb92cb23cf380fed7a : TIMED_OUT in 1h 13m 46s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/6da3738bd2be4692aa51f1453524495a : SUCCESS in 50m 44s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/de2f2a7e31814050a12d45f45dd30d99 : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":1},{"id":"07a462a1af71db46367a1ad80141eae478054ec8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-22 05:26:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"79b470610b89da79418ab20ae9ce15b5902149a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 06:33:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b20da62efc32469fb96749e4ee53372b : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/666870462b634d76a9d71ea1ecaabc98 : SUCCESS in 16m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/504f12cf730a4a68b780bd88486f3353 : SUCCESS in 4m 45s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/341711f41f9649b29414fb2c5edd2dbf : SUCCESS in 48m 32s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/96144edda1064306b7ec39dd5e3bba92 : SUCCESS in 48m 54s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/e72ce5c243784bbfb7906173205ef8e1 : SUCCESS in 1h 03m 59s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/cbca96cce5d24d67a7c09b4d93cf0c79 : SUCCESS in 34m 28s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/70b6e66c54cc4dd5b9aaf7bf96111086 : FAILURE in 45m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ef40c6b522c0431f8c06d9506cf6f30bca5defc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 06:33:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d43d51def8b8bbd525de5e4fb0b38d2659f77566","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 07:58:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f63e9de8610442bb8b09c079356eeb7 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13851c0229c949328c5698ba3a378e22 : SUCCESS in 18m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d841b3404c0c406a89384e1b9f1394a8 : SUCCESS in 4m 58s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d8370251970944dcab93a6ef731950d4 : TIMED_OUT in 1h 14m 24s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/72ad96bd20974f9095ba90ff0e68cad5 : SUCCESS in 53m 03s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/8a6745960191452d853a2bf4d15a8a56 : SUCCESS in 1h 07m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cac51535332220832de3d693b21a0cc6bc724e9a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-22 14:11:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a3c47a811fe8f1bdf10378aa54098cfeff97a65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 15:27:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3dc865b14f4b6bb18485d0efff7281 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bdededde6eb490a96598dbfa238bf79 : SUCCESS in 15m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72a4820492034bccb5fc5b31e11863f6 : SUCCESS in 4m 14s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/6f2614c5d68b4be4b931d52f238f5b42 : SUCCESS in 40m 30s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/ef9384dc4cf244fd90350e37b4a1b72b : SUCCESS in 59m 00s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/30619e687d32413ca686723ecfb5d64d : SUCCESS in 59m 13s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/27487921e5ab443bbe30908057a647b8 : FAILURE in 46m 46s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/444896c6eeb24cc89cc9eae3bb708a43 : SUCCESS in 45m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d22a11303f611a6487a69280ec0cfff7f370531","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 15:29:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81305f40ba95d8dd72d52e48390559b7f7c721d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 16:39:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7bc014b6ee54470a1ee91107416d229 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6073da17091d44ccaea8a91bc64a910f : SUCCESS in 15m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1aa88b6955df4c65a6cd208d46aaadbd : SUCCESS in 4m 03s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/5099704b1bd247d9abe37d68f779ef70 : SUCCESS in 51m 18s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/73f679fc27494316acaf8cdc586d1abd : SUCCESS in 53m 18s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2805f3ce6a894b1d83ecb8b1cfa7cdab : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4c931c4a4bfe80515fe8417f8f16986ace3994ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 16:39:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ad14049a4eab8b6a01ee68fe4b1ceb7af50336eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 16:42:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf179458e0a45a8b946ddbbb91758d4 : SUCCESS in 1m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a366bfeb893491db71e578ab2576693 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40bf033a73c5428a5135e828a0c9967854652114","revisions":{"40bf033a73c5428a5135e828a0c9967854652114":{"kind":"REWORK","_number":1,"created":"2020-09-21 23:44:22.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/92/753192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/92/753192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/92/753192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/92/753192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/92/753192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/92/753192/1"}}},"commit":{"parents":[{"commit":"25ee5972ec0c05da7632988e2d86bc8d343c32c3","subject":"Introduce CNI network driver for docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/25ee5972ec0c05da7632988e2d86bc8d343c32c3"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-21 23:43:52.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-21 23:43:52.000000000","tz":0},"subject":"Add release note for CNI plugin of docker.","message":"Add release note for CNI plugin of docker.\n\nChange-Id: I40e5f01f6df9f318faafe50ee1bf9c91b727859a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/40bf033a73c5428a5135e828a0c9967854652114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/40bf033a73c5428a5135e828a0c9967854652114"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
