)]}'
{"id":"openstack%2Fzun~758652","triplet_id":"openstack%2Fzun~master~I7bbcc2600d775d137c4c3f212aa39c98ff97cab2","project":"openstack/zun","branch":"master","topic":"typo","hashtags":[],"change_id":"I7bbcc2600d775d137c4c3f212aa39c98ff97cab2","subject":"Fix a typo in the explanatory note","status":"ABANDONED","created":"2020-10-17 10:36:00.000000000","updated":"2021-09-21 09:04:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5e91206d35725668809869e2e6590ce23abcc380","_number":758652,"virtual_id_number":758652,"owner":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-17 11:50:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"disliked":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"all":[{"value":1,"date":"2020-11-06 16:23:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-11-17 07:55:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-17 11:50:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-06 16:23:06.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-11-17 07:55:51.000000000","updated_by":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"reviewer":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"state":"REVIEWER"}],"messages":[{"id":"c828586e3915dcf05d31292697f6bb684aebe23a","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2020-10-17 10:36:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27f5307a3da7d930ea83353ebfcdb8bd5a0b6851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 11:50:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c06643ab965461fb7ddd6b95ffb9d27 : SUCCESS in 7m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecd4ff736b174ed0b62641a144021f69 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dec46a514a0344639e4f16a81fcf63d7 : SUCCESS in 9m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/261aeb85f7cd475882e5cd75e63b707b : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4de9758196d14f1a84d7268725ca7d76 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a55862d98343909401daecba94ddfd : SUCCESS in 9m 18s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7460d24595744baea2136b2b7d6201a6 : SUCCESS in 52m 02s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/3806f280195440deaa3200b06126f4a9 : SUCCESS in 50m 41s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ab9e0e23d1c84b4e9a3802d9322d6b60 : SUCCESS in 1h 05m 46s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/a4a02f19751841d3b528fc16953d1e26 : FAILURE in 41m 18s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/7992a37510f3465a84ae7c27da149b05 : SUCCESS in 44m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0c436f08916878064447fce4b4ab2242ba6786b","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-11-06 16:23:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1e5b71d3f41a8a4b1ad6df6fe96a7c33e1319947","author":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"date":"2020-11-17 07:55:51.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease find more in the project and add into the same commit, It will be easier to review and merge rather than for a single word character.","accounts_in_message":[],"_revision_number":1},{"id":"5e91206d35725668809869e2e6590ce23abcc380","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2021-09-21 09:04:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8dfd5b21dce7d7f7f058edb44ddda7b9015da8d6","revisions":{"8dfd5b21dce7d7f7f058edb44ddda7b9015da8d6":{"kind":"REWORK","_number":1,"created":"2020-10-17 10:36:00.000000000","uploader":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"ref":"refs/changes/52/758652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/52/758652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/52/758652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/52/758652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/52/758652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/52/758652/1"}}},"commit":{"parents":[{"commit":"0a38b12573d82a8baf408143b76bbf8043daf14e","subject":"Merge \"[goal] Migrate testing to ubuntu focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0a38b12573d82a8baf408143b76bbf8043daf14e"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-10-17 10:28:41.000000000","tz":480},"committer":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-10-17 10:34:45.000000000","tz":480},"subject":"Fix a typo in the explanatory note","message":"Fix a typo in the explanatory note\n\nThis patch changes \"dont\" to \"don\u0027t\" in\nthe explanatory note\n\nChange-Id: I7bbcc2600d775d137c4c3f212aa39c98ff97cab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8dfd5b21dce7d7f7f058edb44ddda7b9015da8d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8dfd5b21dce7d7f7f058edb44ddda7b9015da8d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
