)]}'
{"id":"openstack%2Fzun~768528","triplet_id":"openstack%2Fzun~master~I3bf267e0b957a5b70ed2e5152d64f97b9880debd","project":"openstack/zun","branch":"master","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-08-29 20:01:31.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3bf267e0b957a5b70ed2e5152d64f97b9880debd","subject":"[DNM] Test CI","status":"ABANDONED","created":"2020-12-26 09:58:16.000000000","updated":"2026-08-29 20:01:31.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eb7c0b26ff3df19ccca008ffb44f4528ef4c45ef","_number":768528,"virtual_id_number":768528,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-07 07:33:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-26 11:06:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acea76f4f84ab81e9a3cd9e98070f6cb36d551e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-12-26 09:58:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35c5d7445d488b2116a101cef529e60cb233e914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 11:06:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/49df31d5b96a4e5c939d3152dcf902bc : SUCCESS in 1m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60a386e01de343449738903b04eca9f0 : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e47c83a7778a497297f479be1e8ce54d : FAILURE in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e527108de224f56bc2a8ef1f84bee12 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/599b013a66aa4cc7a2fba238a12b046f : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e594997cf6694b15a818fd6da44d8566 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1911d199264f4f6c9768fde11e82bf06 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/604f373787d149b49e00ec9f34885ee3 : SUCCESS in 5m 36s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9b9dd28d458f4f2d89d860118ff5d078 : SUCCESS in 50m 10s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/0a4a176a04634057be256770222054ca : SUCCESS in 50m 25s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/6ca3a5a7a163446aaa8e81ef7b2d2e60 : SUCCESS in 1h 03m 56s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/70dcf08384864a37a8b768636404423b : FAILURE in 43m 12s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/3085b501db5a40c2946d2a64cb1a408e : SUCCESS in 43m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67cb1b3683a8fe117bde68815983d4afad84c853","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-12-29 10:42:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"c9f6609c9a0794d4854f3f444b3764e14bff7cd8","tag":"autogenerated:gerrit:restore","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-08-29 03:34:01.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"27ffb0f66c95ecad0ee93b198c331723f2de466e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-08-29 03:34:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"65513a3e493c44ba9f0ee101b30341a5fda5aa97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 04:43:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7673c9e992144f3e8e9170bec7808770 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fab73c5d4f4945b0aecab7037ebc43bd : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51d518b1e6bd4d46a89020d5398d47a1 : FAILURE in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3643d2249947441f9c7cb2756441c735 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b08e9819c17649af98fac87d1fc6b9bd : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99cc15428c9544538011ddd8a0e815cd : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/651675de759a40a3bb1f2182bfa3e0e3 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e3aa3a2ae3458bbe51ec552a9cc90d : SUCCESS in 6m 31s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/85a16b5f7dc9495cab675e9415b9d4dd : FAILURE in 52m 21s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e01b32f656cd4b00a048651c57360894 : FAILURE in 51m 35s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/3f590a6b2e134d7a98789a0da59525a6 : FAILURE in 1h 03m 44s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/9c59111a2d6a418388296b1a76adff23 : FAILURE in 37m 15s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/5551f8876e9a411d9a3c8afcd718c1ab : FAILURE in 49m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18fd193a1707ce92970c1cfd65282f7f49c43762","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-08-30 03:39:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"21705b420a6ce919dd4d7586162b73c2e5035861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 04:51:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4203d935a16b4e9397582dd86a10c8b0 : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9015b0b2f6054134b8b5bc9959a2f52a : SUCCESS in 8m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18dd124e333a4fc6a147c590f0369f23 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9ebcaabb2ed4dec874cab62d284de58 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ead53a9525b14868b3fbcad991086251 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3278dd159de44f7901045d68999aff5 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb4a061a1dfd4dd8aaf6758723b9dd48 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/911874ca36ea4db7aba065e456a9dbce : SUCCESS in 7m 25s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/f086623e08f7418ea5b6d42742cff3d9 : SUCCESS in 52m 49s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/2b9b0b40ae624f3d82da1e7bcae670c8 : SUCCESS in 52m 33s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/a19395447b4546e69e52f63506f90339 : SUCCESS in 1h 04m 42s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/48fda4942f2343fcb86a6f4169311003 : FAILURE in 35m 31s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/7d695ebcec494bd79872a71ddd93ca83 : SUCCESS in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf4dd3cea34f95a67c407d4411cd1064e2b509bc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-10-04 04:19:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"b6061962dee088ef84db359b28a92813fc2e8c26","tag":"autogenerated:gerrit:restore","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-04-25 13:39:27.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"098970c799a429bb743d465462896699da4029b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-04-25 13:39:39.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8d9766a60898acb5d56ae48efb8bcb5cd60a64ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-04-25 13:41:01.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a9b08049c3e4f77787aad78feebfd961b97728a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:59:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8749ecc6f77d4f11b354a09e7b9bb288 : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/609f3808d0074db7856dd8d01e1fb271 : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04fee07fe2064be7a895cbba7cf84bd6 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df6bdea62f374a95bc994835e6335815 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b81d64bc36cd4c039c6b191905f5045a : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1c53384247e4b2dace04fb2c524a973 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca701605ea847cd8fa1af399ad93c51 : SUCCESS in 8m 09s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/cfbc898248114f41b2032e05bf566f49 : SUCCESS in 57m 57s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/5d329ed9811b4fbc8b3fee0d3d522b1c : SUCCESS in 54m 09s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/6fbab28da823475fb931563676c53cf1 : SUCCESS in 1h 10m 35s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/2c258853ccb54b98bd13ed9a36d37184 : FAILURE in 45m 41s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/143e2734f57b4d99b81eea79cbb21570 : SUCCESS in 52m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99c4d48b33444cc5836f49275be9d8f6b5b07631","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-04-26 06:59:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"50cf6e2f3f4c823bafc9c2ea0cccaefe2e49f479","tag":"autogenerated:gerrit:restore","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-07-31 09:30:09.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"de7685b00928578e2c50eaac6f8e4478b44fb232","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-07-31 09:30:17.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2ae834948bb16cd6d0a69a93207e88f6e3387044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-31 10:51:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dba8b4083064e78b4ca074197e1100b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/db30d299643d495783f8156b388ece95 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14327904cedc44f0a59fe529710981ee : FAILURE in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/248e05fa31a64c0d9bf686b735c907f3 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3209d8d58dfd469a88bb6d96f7873511 : FAILURE in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04bcc2d66eab4707869ad1fb48bbaca0 : FAILURE in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a96811d2389c42e6953cf9212e305c5c : FAILURE in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1257c4db6d24641a1d11be052b04ea8 : SUCCESS in 6m 50s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/4af49f94673f445da6d2f70af922bbb0 : FAILURE in 24m 09s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/851578beebc2467d8554b6606267e230 : FAILURE in 24m 34s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/25015d7735db4f4fb48bf940c8311e3a : FAILURE in 29m 41s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/7dedc0c9faad40999d424c12b2f86fee : FAILURE in 25m 45s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/3ee4b154966b417f8ee65671bd0e71e0 : TIMED_OUT in 1h 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1e48d5cc5dc8e55153d60e934674fa092e06f6b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-12-09 13:46:10.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"4ab3dbd97c491842b0d069ae84555ef2682716e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 14:56:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cd0b3eeffe94e19a667469fd5d98078\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f8f61d055bd746dc871cd3292dcdb927 : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e006e37416ea437fbe8c0f78fe9e2ffb : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31f1604ebf2c4296901878b4b2e3f330 : FAILURE in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0969a0c938a64c1793a0dd60b0571a16 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6094e569b92547229d4008f47429b56a : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94263c2325db4ceb9047b177c05e8452 : SUCCESS in 6m 03s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/fb4838bcb939434389594741bfc3ad60 : FAILURE in 1h 01m 51s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/9cdb6aee1e8a4c6aa8972e2cf1ae3724 : SUCCESS in 34m 04s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/9234eacfb14f457c9c8986ba11a03b4c : FAILURE in 37m 01s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/bca4a39ae63b48529aa62e746470277a : FAILURE in 22m 16s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/369182ddbe774012bc135c13b8379f57 : FAILURE in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bd3c2aa28f0267d16e27d5d98eb907cc63d2c229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-09-17 23:44:10.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"78e807b8794a2826dc61cc586d1934b856a02384","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-09-17 23:44:27.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"7068356a3a84c5b420c05275574b32113c25b3b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 02:02:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5b7abf465f84bcfadc657b1da05920c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9fec747f5a3e4c14858dbc764ca2171e : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c2f0821497a4bd3bd1c26c14ce2a990 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518bc59cd2f64b579dd70a917baa8999 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/771471294f13462dbbebb297afc0c6b8 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5c9c0ddb93649e59c03fe4a114f86c2 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d465bcee8aaf4f22b0c4cacfd6a410b4 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c69a16c3732e474c85d7ee6bdc995fad : RETRY_LIMIT in 1m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46faf55a9ba54d279276f860c2f08aa4 : SUCCESS in 8m 23s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/01f29b4cc78745d7880c79332567ea65 : SUCCESS in 52m 46s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/3e29d3fa9ae04208be50299f142d7c23 : SUCCESS in 47m 09s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ae80ce62b7e440bc9af2a748fcfea8fc : TIMED_OUT in 2h 12m 46s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/81fcae78df7d41018ae7186d547165ad : SUCCESS in 21m 35s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/d66c808e5691441e85d5bb300d3a8c37 : SUCCESS in 48m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2636929884cd5ee90737e2cb99aa7c6948ee4bb4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-09-18 13:08:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a109e9d0cd478614222ea2661d3e18f888c03530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 15:26:20.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b93bbd39b53c4049b264bfe9356f86fc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5fdb31ede1484ceaa45d899c0c8cd3cf : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/774ca83b52204fe4bb2e8bca736b22a5 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30849987361944308583bf11b5737e79 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a04064b058b54e269306a3657472a78b : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68fbf0d58ead4703adcdef71ba0e37ef : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c306038b07db42048939c39f226b7af1 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a0493c4afc44c9da934bf69b32b6462 : RETRY_LIMIT in 2m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aecd1227cd534797a90d6f53b0c4b294 : SUCCESS in 7m 37s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/411c974990e741318191ed96726fa8bb : SUCCESS in 43m 43s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/3d769f925103430a8e763764fd80f981 : SUCCESS in 47m 57s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/3401f10832e64935bbea6ffc183c84bc : TIMED_OUT in 2h 12m 40s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/c0c49d2af7d24d7e9ff8c79dfeaefcf9 : SUCCESS in 35m 22s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/e4cb638b53734d58876207fc5d87efbf : SUCCESS in 47m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"49690276ed6f666055ba37405de6fc54f2bd5a14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-12-18 02:37:45.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"14130d5c8277807672237d683d1955e6dfbb236f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 04:55:24.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd77be84e7c6406ba343f136b9d555ff\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fd0a6d4239ea4949a6e191cc3744aaf9 : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eac0bea3a8f14d9ba520c2181eda2395 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fa82a4d916c4bd8b3f573561e184ef6 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6cb4fdd2ead4b119f472c0a18482df6 : SUCCESS in 6m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78d652b04ef54582a9e50fd59a66a48c : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124924802a1543329966af3466f5ac24 : SUCCESS in 7m 44s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/fc6ffaa720634ccf9f77ac7fd197b825 : SUCCESS in 50m 38s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/27235ea845f14f638a113f7caa8b6f62 : SUCCESS in 46m 22s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/9585fa8282f447dfa4c33ce003237cb3 : TIMED_OUT in 2h 12m 52s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/cbd64fde5d294a49bf4d15c8bd93a249 : SUCCESS in 28m 00s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/2f5efac5798b426cafc6ca8068634a8c : SUCCESS in 47m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"958a7aa2d4398f377848c1a0535edc78f168aba6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-12-19 11:02:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"36b0fc8e2411626a826a43132bc09a82509faadd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 13:21:22.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39964b698f514ef984535d5a65be815f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fd6758c257804b35aebf9fcc4197d137 : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f0c6f88fa1d4d1e8837a80ca554d4e0 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a91ed50f874886a178539741e6fa06 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a4e5d8fd9184a0ea997086442711d33 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31398eeb87584671937bcbf1d2ef8dd8 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28f9469e8db044f9929531790165431f : SUCCESS in 6m 21s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d05979d43678466d8265e774dd2711f4 : SUCCESS in 56m 18s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/f03ce8015266436996702b1278f8c701 : SUCCESS in 37m 17s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/e64bdb58ee67435a80923d3ffdc623ed : TIMED_OUT in 2h 12m 48s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/dc7693159e874a30a15e6d3dbfe7ea7a : SUCCESS in 25m 52s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/61b33c617dd3450398a5dcc3810724f3 : FAILURE in 53m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"bb3e9bbf5491e79b7b38e99505c5f7c02d9aa975","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-04-29 05:53:23.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"08e361060a7665a1f234def4dc3a0eb232673e06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 06:48:46.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/121855da852542dda0f5a398ef29d1f0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ddc8fca8df3343d4a0413844cf6ad70d : SUCCESS in 1m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12643c96b05b4a3c87ed2a39dd824864 : FAILURE in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10201b0f04a84246b81c3216cd003d6d : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8135504d0424411d9ceb781ca73558f9 : FAILURE in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e188d9dfde4247678c05e4791a61b54f : FAILURE in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9acc6360b99b4b57aedc91178207b56b : FAILURE in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5701feb5d04899a63f5e8de1938591 : SUCCESS in 7m 42s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/390eff07e4174ecc8c8c3be164ad55b6 : FAILURE in 46m 20s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/d823f8803bc644d3946195b0195ff870 : FAILURE in 40m 34s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d0d4d865264d4edf834f65890af22966 : FAILURE in 40m 32s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/f8c1c21c23dd441bb856e66a2f195da4 : FAILURE in 29m 20s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/e050381031e840d8ad34374246fe11e5 : FAILURE in 49m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9fb4f5e9c64f65bcb6d0acfcf92ed0792deb8694","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-12-05 10:08:05.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"06eb7534504df653c371d3a3c7c78412eb693ced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 10:54:20.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4993a30d2f7043c080e1f003add4bf15\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ca8fc0ebcead441e871982e9f97108dd : FAILURE in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f5145dde1ed4fbea01953dd9388843d : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3da12097f644eaaad21abe352debe0e : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a70dca0b814e4957aad76f8b5534e4d0 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eee9d6673c4d4bd7a846488e31e4914c : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/858bbcfb802c42428ba9ca553538ac4f : SUCCESS in 6m 31s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/036588fbb6774d3b94173026e6649c3f : FAILURE in 37m 34s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/d9f19bdcc1634246b9a79d32afb82b3c : FAILURE in 38m 43s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f463a95646a94a52a9dc187624ab9691 : FAILURE in 43m 46s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/491e522109b04b69ae54d11f71144d39 : FAILURE in 32m 49s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/c33710f5051740088d0bce9ed99feced : FAILURE in 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"96bfc808303848d959562db80e85fcab7c64ea4e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-12-06 02:58:01.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"be25dd3eabb7114bc4b45c78a9fcd41bae9b90dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 04:31:12.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d50e6225bbae4947969ccde0e7241306\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7fdeb450f23494b8c5bb24830e48d6b : SUCCESS in 3m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bde885ada7f142d980f7caf70b104207 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b137ac9ee274d1aacf8b90717e3cce8 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bf68b5d16da41eb83b4665cd9f2d2e0 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b28b533b6e924ad9a078ee00f4abb5a7 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df76191d9fe44e80bef59619d64d91cc : SUCCESS in 6m 24s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9c43e17f40ab4da2bd6fdcec3e9ab94c : FAILURE in 45m 20s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/0010a36912e6424ba67156a3e60c7f34 : FAILURE in 46m 53s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/bd43b34577c24825acd6caea311d9952 : FAILURE in 45m 29s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/af3896247d5641ee98ec62c129fc7358 : FAILURE in 38m 47s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/afb2540f0bf740f7b2abe9894d056147 : FAILURE in 27m 42s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3e22826a20255bb5c2ef9ed5fefbf5ebf5aeb3e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-30 11:21:21.000000000","message":"Patch Set 12: Patch Set 11 was rebased on behalf of \u003cGERRIT_ACCOUNT_11536\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}],"_revision_number":12},{"id":"b18a5e9b69a82956d471be493a38158cc276740a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 11:56:29.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94852968d46c412b813c1f0abe4a4e58\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9dd32386059c4006a83962bc04eca181 : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1987a717774f4e55b941609a006344ad : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59dc7823dec44cdbbbc25cfd8c64d0b4 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9b1ad78d0f4482c9a5dcfb4ea822dff : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2b865e7f8994e319a1090ee065d1cf8 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4fc573d5f4e4b5c83e1b3cf20ac4509 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c68f03ae6c84c1aadc262bfb6b28a72 : SUCCESS in 8m 12s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/67bdab9a94644e47955a8f3fc9b4f370 : FAILURE in 12m 23s\n- zun-tempest-py3-docker-sql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/320aab5c22b1444f8b889a05f884dbba : FAILURE in 27m 35s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/03a79071e70a4cddb07df020b7f6ec21 : FAILURE in 30m 30s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/79cdc29d696f4feca686688cf44e456e : FAILURE in 30m 20s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/54ccc63c403e477fbd89c8298707f606 : FAILURE in 11m 38s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/b48f64a955834c00a6a0f513a165d341 : FAILURE in 22m 37s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c81bd7a9391a1b1dc32f84896e8753fe021dfe70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-09-07 06:59:27.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"8ec1d26de8db6b75232fcf9c9a52018b6b444697","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-09-07 06:59:35.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"41c5fd81bf6b983dd3f90a96d83b1047b1245220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-07 07:33:27.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28a081b38d4c4b998e3f73938f8cdbf5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c75d922c22ac4bfdb11e1dbf33337254 : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/527879b34d9540feb6803334fe5662da : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac3af51d0b74d75821b67a4467fc79d : FAILURE in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2edf81dc510c4d209ecebbe191dab38d : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/563789e719334a1699146d297fe5f58a : SUCCESS in 6m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc1052f60dcc4b27ab7df53d2ddabb70 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0966e155210409382cd17980c687f5b : SUCCESS in 4m 29s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d2819c90b63842d39e7b32ebc6af6fae : FAILURE in 12m 45s\n- zun-tempest-py3-docker-sql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/476aa007d3414154a4cda5da349a18ba : FAILURE in 25m 26s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/b12373c4c58d4bbf99cc98d5b7e5de08 : FAILURE in 26m 08s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/e3e4ee90c1e7458fb9b48e26aff5a4cb : FAILURE in 27m 02s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/5f9f93a6aeeb47aa8a386fd4347c4bad : FAILURE in 17m 54s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/432dbf260c5a400f86b075c3bc642fc5 : FAILURE in 23m 32s (non-voting)\n\nWarning:\n  Comments left for invalid file contrib/nova-docker/nova/virt/zun/driver.py\n  Comments left for invalid file contrib/nova-docker/nova/virt/zun/opencontrail.py\n  Comments left for invalid file zun/api/http_error.py\n  Comments left for invalid file zun/api/middleware/parsable_error.py\n  Comments left for invalid file zun/cmd/__init__.py\n  Comments left for invalid file zun/common/utils.py\n  Comments left for invalid file zun/common/yamlutils.py\n  Comments left for invalid file zun/container/docker/driver.py\n  Comments left for invalid file zun/db/sqlalchemy/models.py\n  Comments left for invalid file zun/objects/fields.py\n  Comments left for invalid file zun/scheduler/client/report.py\n  Comments left for invalid file zun/websocket/websocketproxy.py","accounts_in_message":[],"_revision_number":14},{"id":"eb7c0b26ff3df19ccca008ffb44f4528ef4c45ef","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 20:01:31.000000000","message":"Abandoned\n\nHello, CI now fine.","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"7ff00ca339ef0e35216c3fa6cac1b9994965cf58","revisions":{"209e9f332937937f1c0975ec53cae7c52d1d27c3":{"kind":"REWORK","_number":1,"created":"2020-12-26 09:58:16.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/1"}}},"commit":{"parents":[{"commit":"12fa7efcfef617a7112c2c017f744767c56d212a","subject":"Fix invalid argument formatting in exception messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/12fa7efcfef617a7112c2c017f744767c56d212a"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:58:13.000000000","tz":480},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/209e9f332937937f1c0975ec53cae7c52d1d27c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/209e9f332937937f1c0975ec53cae7c52d1d27c3"}]},"branch":"refs/heads/master"},"8152f8df9ca33ef17bb655bf0a7c20a923b4e01f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-29 03:34:13.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/2"}}},"commit":{"parents":[{"commit":"abd1e5c9f34c3b3c141600f0386e77ab41f7e89a","subject":"Respect allocation ratio on container update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/abd1e5c9f34c3b3c141600f0386e77ab41f7e89a"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2021-08-29 03:34:13.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8152f8df9ca33ef17bb655bf0a7c20a923b4e01f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8152f8df9ca33ef17bb655bf0a7c20a923b4e01f"}]},"branch":"refs/heads/master","description":"Rebase"},"af7b704815ccce9ff9baac944fd63f7531b765a7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-25 13:39:39.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/3"}}},"commit":{"parents":[{"commit":"2e8825a05f397b01b760670d606f3c3cb2a8d3cc","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2e8825a05f397b01b760670d606f3c3cb2a8d3cc"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2022-04-25 13:39:39.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/af7b704815ccce9ff9baac944fd63f7531b765a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/af7b704815ccce9ff9baac944fd63f7531b765a7"}]},"branch":"refs/heads/master","description":"Rebase"},"7b6c4ad3074504819665abb43fa147347407e9b1":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-04-25 13:41:01.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/4"}}},"commit":{"parents":[{"commit":"2e8825a05f397b01b760670d606f3c3cb2a8d3cc","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2e8825a05f397b01b760670d606f3c3cb2a8d3cc"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2022-04-25 13:41:01.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-libnetwork/+/839216\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7b6c4ad3074504819665abb43fa147347407e9b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7b6c4ad3074504819665abb43fa147347407e9b1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2b4d47832ece2bdd9b79696d50a549eeda5ee57b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-07-31 09:30:17.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/5"}}},"commit":{"parents":[{"commit":"a4e5688a68e48cb4dd066c49a2d7856de3bba92b","subject":"Drop lower constraint job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a4e5688a68e48cb4dd066c49a2d7856de3bba92b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2022-07-31 09:30:17.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-libnetwork/+/839216\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2b4d47832ece2bdd9b79696d50a549eeda5ee57b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2b4d47832ece2bdd9b79696d50a549eeda5ee57b"}]},"branch":"refs/heads/master","description":"Rebase"},"df38f2a3b3e5fba3b1a537d8e73033330ab015bc":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-12-09 13:46:10.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/6"}}},"commit":{"parents":[{"commit":"f7f67cbf758e8c0b7dd9387b1607fa2f3d745c78","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f7f67cbf758e8c0b7dd9387b1607fa2f3d745c78"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2022-12-09 13:46:10.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-libnetwork/+/839216\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/df38f2a3b3e5fba3b1a537d8e73033330ab015bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/df38f2a3b3e5fba3b1a537d8e73033330ab015bc"}]},"branch":"refs/heads/master","description":"Rebase"},"5583570aca98855060967cae4f5010ff4592cff1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-09-17 23:44:10.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/7"}}},"commit":{"parents":[{"commit":"f7a526b107865a3c86fe2c9474b616c4ecc497ab","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f7a526b107865a3c86fe2c9474b616c4ecc497ab"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2023-09-17 23:44:10.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-libnetwork/+/839216\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5583570aca98855060967cae4f5010ff4592cff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5583570aca98855060967cae4f5010ff4592cff1"}]},"branch":"refs/heads/master","description":"Rebase"},"fb8bd2ca59338b3061e22833f154c7eaae5c21cf":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2023-09-17 23:44:27.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/8"}}},"commit":{"parents":[{"commit":"f7a526b107865a3c86fe2c9474b616c4ecc497ab","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f7a526b107865a3c86fe2c9474b616c4ecc497ab"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2023-09-17 23:44:26.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-libnetwork/+/895578\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/fb8bd2ca59338b3061e22833f154c7eaae5c21cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/fb8bd2ca59338b3061e22833f154c7eaae5c21cf"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8d2361145b32601add2b1b043508cba2be125a54":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-12-18 02:37:45.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/9"}}},"commit":{"parents":[{"commit":"397edf53c161e9dbc9b26ab152227142d8c38796","subject":"Use alembic alter_column for 35cb52c5553f migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/397edf53c161e9dbc9b26ab152227142d8c38796"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2023-12-18 02:37:45.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-libnetwork/+/895578\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8d2361145b32601add2b1b043508cba2be125a54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8d2361145b32601add2b1b043508cba2be125a54"}]},"branch":"refs/heads/master","description":"Rebase"},"951bbc5b09c90d4daa76444d9b8ac604cb209ae3":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-04-29 05:53:23.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/10"}}},"commit":{"parents":[{"commit":"16eb87ee1cb5f04fd32d7277b4d6455b71925d24","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/16eb87ee1cb5f04fd32d7277b4d6455b71925d24"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2024-04-29 05:53:23.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-libnetwork/+/895578\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/951bbc5b09c90d4daa76444d9b8ac604cb209ae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/951bbc5b09c90d4daa76444d9b8ac604cb209ae3"}]},"branch":"refs/heads/master","description":"Rebase"},"59b9f154193be77e0e0bbde13600e045ff9dc932":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-12-05 10:08:05.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/11"}}},"commit":{"parents":[{"commit":"5c5e1063edb452418cfe5b9dc452016171738312","subject":"Remove workaround for eventlet bug #592","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5c5e1063edb452418cfe5b9dc452016171738312"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2024-12-05 10:08:05.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-libnetwork/+/895578\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/59b9f154193be77e0e0bbde13600e045ff9dc932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/59b9f154193be77e0e0bbde13600e045ff9dc932"}]},"branch":"refs/heads/master","description":"Rebase"},"24b48b609d748538644471d087fd7a33f63beb5d":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-04-30 11:21:21.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/768528/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/12"}}},"commit":{"parents":[{"commit":"eee3c483117e30e261951cbd4d58fdb8e4f5a832","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/eee3c483117e30e261951cbd4d58fdb8e4f5a832"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2025-04-30 11:21:21.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-libnetwork/+/895578\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/24b48b609d748538644471d087fd7a33f63beb5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/24b48b609d748538644471d087fd7a33f63beb5d"}]},"branch":"refs/heads/master","description":"Rebase"},"9e58a34fe2197eb82feef6b5d4f2394fa103522f":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2025-09-07 06:59:27.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/13"}}},"commit":{"parents":[{"commit":"eee3c483117e30e261951cbd4d58fdb8e4f5a832","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/eee3c483117e30e261951cbd4d58fdb8e4f5a832"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2025-09-07 06:59:27.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9e58a34fe2197eb82feef6b5d4f2394fa103522f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9e58a34fe2197eb82feef6b5d4f2394fa103522f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7ff00ca339ef0e35216c3fa6cac1b9994965cf58":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-09-07 06:59:35.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/768528/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/768528/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/768528/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/768528/14"}}},"commit":{"parents":[{"commit":"023280e4227d78d56c4d511fe8ac647423fd9d65","subject":"Merge \"Zuul: do not use USE_PYTHON3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/023280e4227d78d56c4d511fe8ac647423fd9d65"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 09:57:22.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2025-09-07 06:59:35.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nChange-Id: I3bf267e0b957a5b70ed2e5152d64f97b9880debd\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7ff00ca339ef0e35216c3fa6cac1b9994965cf58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7ff00ca339ef0e35216c3fa6cac1b9994965cf58"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
