)]}'
{"id":"openstack%2Fzun~769864","triplet_id":"openstack%2Fzun~master~Ia1b64a4a66a876e38ed2885b543df2acc24a379d","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ia1b64a4a66a876e38ed2885b543df2acc24a379d","subject":"Remove __unicode__() from ZunException","status":"MERGED","created":"2021-01-08 02:58:18.000000000","updated":"2021-01-08 10:38:23.000000000","submitted":"2021-01-08 10:36:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769864","meta_rev_id":"47cf081a80bf666fab58787579f1ab8f2dd0e9d6","_number":769864,"virtual_id_number":769864,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 10:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2021-01-08 09:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2021-01-08 09:17:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-08 04:11:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-08 09:17:19.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"96bcebf12671195180826bdfc8b20763bb17164c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-01-08 02:58:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c825af480c35c8462bc3229033fa077b6a93af3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 04:11:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1840ccd175d843059f9c07667f51945f : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9d37820b0244deab8c0c1d72fe15ba9 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9bd2d54c4c0474daf669706d612a6e5 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6e38ea08ad44d35b06d0d60f2e33e25 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8adecfb13b240dbb027ccf4cee86983 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f4496080d314bac862a239de172efe2 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417ea1859ffa417296e762ac0d40580f : SUCCESS in 5m 45s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/50cac06e2046410aa292ee4128935222 : SUCCESS in 48m 08s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/80c97b82fbd447918a310404c76292f5 : SUCCESS in 47m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/e9757452b92e460096810272d9229171 : SUCCESS in 1h 07m 48s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/eeda0953e4694e34ab83099a2be38e98 : FAILURE in 41m 33s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/808c554e22e84d32b9110ed611882f9b : SUCCESS in 43m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f7852994626099e46bb2327098a8d0189c3a35a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-01-08 09:17:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c89082b4d7aba827c4f54970553efccdc352e28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 09:17:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"983ef8e11c1d39eb9c09053fe267ede51a6cbcb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 10:36:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46e88f873a384d9fb74929a2e74797f5 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a64d4b0ea0249bc9c5937c0ace5a951 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7191a1c9fcb544d18e76457c4084a273 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b97655f80f544a58368b98b01ca6595 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af96c91a95c48dda024e36b80f2a0c5 : SUCCESS in 5m 47s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a66477a104c549fc8f84c147fd727ee4 : SUCCESS in 53m 57s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/97f446d680ab47aea2ace3cfa1829983 : SUCCESS in 51m 54s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/26130ebf096843538eeefd1eb58d0711 : SUCCESS in 1h 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"dbbfa2ccce7830c363a8fe45d797314b03583e09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 10:36:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"47cf081a80bf666fab58787579f1ab8f2dd0e9d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 10:38:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04106e28dc644e70a20647c20d0c90af : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ac7a3ee5322164995248f4534f807e897991180","revisions":{"6ac7a3ee5322164995248f4534f807e897991180":{"kind":"REWORK","_number":1,"created":"2021-01-08 02:58:18.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/64/769864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/64/769864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/64/769864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/64/769864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/64/769864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/64/769864/1"}}},"commit":{"parents":[{"commit":"8e7e4eb98608ecaa362b4d442b6550585f69112f","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8e7e4eb98608ecaa362b4d442b6550585f69112f"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-01-08 02:57:23.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-01-08 02:57:28.000000000","tz":480},"subject":"Remove __unicode__() from ZunException","message":"Remove __unicode__() from ZunException\n\nThis is no longer needed in Python 3.\n\nChange-Id: Ia1b64a4a66a876e38ed2885b543df2acc24a379d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6ac7a3ee5322164995248f4534f807e897991180"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6ac7a3ee5322164995248f4534f807e897991180"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
