)]}'
{"id":"openstack%2Fzun~798971","triplet_id":"openstack%2Fzun~master~Iec820083b26a6901bd582c32bb6ad825c47ca90a","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Iec820083b26a6901bd582c32bb6ad825c47ca90a","subject":"[DO NOT MERGE] Test CI","status":"ABANDONED","created":"2021-07-01 06:26:10.000000000","updated":"2021-07-02 02:41:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f5bfb07782e5d780f3ebf61dc06d721bb5fcdc2c","_number":798971,"virtual_id_number":798971,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-01 07:22:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-01 07:22:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7d0b18bf4fb57bc01341d8935a83c603f4cebfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-07-01 06:26:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cde55a51f4d5adc603da271f19df85e441b5ec2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 07:22:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/95ca7b03187548cbb55d49377c68ae3b : SUCCESS in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2507cbf59adb4f839987a0319e448f21 : FAILURE in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c7fba28a6944f9c925d64435625dfa6 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dab8e98d37124eb18f014ab4d930a64c : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3526e6d868a478d89aab86306f120c2 : FAILURE in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50db7d3f13034d2fad27124b6d76e0ac : FAILURE in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50d7f41114e34acda1140e44598da700 : FAILURE in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1ba09d4075c4e9ab0be5d5b158147fa : SUCCESS in 8m 28s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/e2d97e3c2fc04d17a73b12b211564cb8 : FAILURE in 34m 43s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/d2275878b3f9431b8b3c5c62d4aba92d : FAILURE in 39m 44s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/9440cdca227c4ebfa983bf28122416df : FAILURE in 53m 32s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/9387d4532f7a4ba8b16d9fe95b8e234a : FAILURE in 35m 29s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/5943790ed05749bc9bc16411c84ec450 : FAILURE in 30m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5bfb07782e5d780f3ebf61dc06d721bb5fcdc2c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-07-02 02:41:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"464d235f03d93f7fddce4e924604cc35e3b8a5a5","revisions":{"464d235f03d93f7fddce4e924604cc35e3b8a5a5":{"kind":"REWORK","_number":1,"created":"2021-07-01 06:26:10.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/71/798971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/71/798971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/798971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/798971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/798971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/71/798971/1"}}},"commit":{"parents":[{"commit":"52a35609b0a91857d1119a03c2882e4cd28cf52d","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/52a35609b0a91857d1119a03c2882e4cd28cf52d"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-07-01 06:23:44.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-07-01 06:26:07.000000000","tz":480},"subject":"[DO NOT MERGE] Test CI","message":"[DO NOT MERGE] Test CI\n\nChange-Id: Iec820083b26a6901bd582c32bb6ad825c47ca90a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/464d235f03d93f7fddce4e924604cc35e3b8a5a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/464d235f03d93f7fddce4e924604cc35e3b8a5a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
