)]}'
{"id":"openstack%2Fzun~809082","triplet_id":"openstack%2Fzun~master~I86bcd30139720be4c5949a69075efefaa737990c","project":"openstack/zun","branch":"master","topic":"add-yoga-python-jobtemplates","hashtags":[],"change_id":"I86bcd30139720be4c5949a69075efefaa737990c","subject":"Add Python3 yoga unit tests","status":"MERGED","created":"2021-09-15 02:11:36.000000000","updated":"2021-10-04 05:26:59.000000000","submitted":"2021-10-04 05:22:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809082-add-yoga-python-jobtemplates","meta_rev_id":"5f0a4c60709daa5c2ff634db7dd769ad28385c50","_number":809082,"virtual_id_number":809082,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-04 05:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2021-10-04 04:17:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2021-10-04 04:17:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 03:29:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 12:53:27.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"CC"},{"updated":"2021-10-04 04:17:31.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"53a5da9e4c10d963951a1e8aa216aed91857c3d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-15 02:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"145a3033b3e69417996550cb7d9d6f86228952a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 03:29:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2095d112b89a429197be080654e522b7 : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aefcf9c4c96a4588b84052e6e7104507 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bde69ce0f0748ef8b4438580e964bb9 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ef696ebebf049c29f79f1203fd8a22c : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9344b9056f624c4280e49e517f35c450 : SUCCESS in 8m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7577b4c408e438c9f9e6bf3ffb453fa : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4088eb4d8b4d30be1f42bfc2055c91 : SUCCESS in 9m 48s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b36091bf8eab4b5eb47900ec1816dfe3 : FAILURE in 45m 53s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/09973f6d1dfd417a947dd07ed502b607 : FAILURE in 21m 17s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/4058bd8e662b4ebbb09dffe84b9dac6c : FAILURE in 1h 01m 43s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/5865ba2eda604e18ad07b9d0dd6f4500 : FAILURE in 39m 15s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/cdb3b510c2c5421099bf8be966cb533d : FAILURE in 41m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d45b559bebb6f8d58d601843fa334f2aa2199310","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-09-15 12:53:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"deadec08b04a8815333ef3b7906b28b8e711a39e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 14:01:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cabacb1c4db4a0d8a6d2bd01f56f21d : SUCCESS in 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1941ff86255e4968a6c0eac9c2e3cd96 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16585e0461f1496cbb6332b4a8490957 : SUCCESS in 9m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81491c7a511443f4ba585806727cde00 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ad5a317b70d4e7f90c4b04cdd0f3ff3 : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0346d570063b4f06aa6f5a740ea64805 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4459ff33086042b6aea8aa470641ba58 : SUCCESS in 7m 44s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/502c7e8708fd404c85110141b0c56b26 : FAILURE in 41m 22s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/8e3bf09d13e442a091b58972e1b7104d : FAILURE in 49m 58s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2b7b3b8c5e5a4aff99bc2297c9d8a6b2 : FAILURE in 1h 02m 45s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/2fcd0d02408e489cb569fa787d7d0acf : FAILURE in 30m 56s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/324dae454209499c95800e4186962cff : FAILURE in 36m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87d9152534a1d4f85a41aa0af972fba73876ca00","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-09-16 04:34:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"380f5a84c2ca71a19f489e59e4474aa198827930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 05:30:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e3411f36eef4c049f9456ed6842e8fa : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/640e16bfa2c74cee9c44313eeb94a848 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8a6ae54513943d8b730a1bed15f86aa : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c493a1730a354b07bbff91c562806bb3 : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/712f97b965bc4486a71a9c898264e31c : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc206cfc2d914785b6d6989520ea8225 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c33a1e35ec40b5bc77538b7425605a : SUCCESS in 8m 00s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/de4ae0a2a55d4f4fb27d2f8b4010135c : FAILURE in 51m 00s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/42c11512dccd484fada8cb41ebbcc84b : FAILURE in 40m 02s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/166c7a50310742debf98945446616a32 : FAILURE in 51m 53s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/3c67df2a139947769ed65cefbb63cc86 : FAILURE in 34m 20s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/11847c92cd584553ab7cf2ddcf78b977 : FAILURE in 38m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eaae9a1130644c66bf09fd6e017af1fb3bb45d0a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-09-16 09:56:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6bef9d081bf9ad7f478ee98aadec23fe511bb197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:57:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/172824d4c17546c68dafe32137a038cd : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81286ef2b8a54ca08f32d5727ceba6c1 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32649cb0b2ee48a2ba950477ee1d5a14 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12410e6064e24a2485e7bad0ff71ff4f : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb9fc7b543bc45d0aad6304587c14d31 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71bbb043fb314c3089f41c99722c3687 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf64e4e6e1f4ba188174c7d88dc8f69 : SUCCESS in 6m 43s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/e760beaed9f448068711d1b4e1994017 : FAILURE in 46m 08s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/4efda7d34e3e4ff2bd3fde40eefb5ce9 : FAILURE in 47m 43s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/21d2a14ef3244979adb240116d6461b9 : FAILURE in 56m 36s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/2a5ebcf9bec044ad875d35a8b9f9aebf : SUCCESS in 32m 43s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/17a7a9b3d72d4cb7821a60134d3bf8e5 : FAILURE in 39m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d76674c7be1e04df74f92416f3f0e93fe6b92365","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-09-23 03:36:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"adafd63a0139ffe7ff4d6e5c0ff2dc44177a9316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 04:37:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b61b34793fd14dfba044174e0bc17b2b : SUCCESS in 8m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5242725cec2b47c5bd59ff900e2ea4cb : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c20c87e671894b87a83d2d6b7463a4a8 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ff85d3c80654c368fca55537faf3ae8 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fc93cd5ee9441e8bfc8444268ec5f3e : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a10b1b4d3b28416994154697d02d8b1b : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ba4b09c8ceb4b0ca9f810292919b49e : SUCCESS in 6m 44s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/c0d68f2c8a18444ebd64c73882c378a4 : FAILURE in 48m 01s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/08352cdc61f3400ab74e991926b57508 : FAILURE in 47m 26s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/21db26b5018b42bc8256bbdd13b440ca : FAILURE in 59m 07s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/95e193ace7e84c90910e98135bb89fef : FAILURE in 38m 57s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/a275c797d724403b85779a7963824396 : FAILURE in 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1011b416908b6acbd84308c00f48fe5f74296c7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-09-26 05:32:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8732d3303cf061401f6b642779f91b5beb8898ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 06:37:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c5c5e2ff3724183a17bc8145dd7ba97 : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b45935a547e4e99a5deb5f954e2d53a : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be58a83ea6c42d1b35d77bc78508583 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65ec05f62fa646c08c3bdca33af4302a : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e520188bb9ee4e789509257087def577 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9023365d5104118a096541f5638223e : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc967387543445ef99a05ed59aa76434 : SUCCESS in 6m 55s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2aec98931e9d4a3fb33ea69ee6b04bcc : FAILURE in 47m 46s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/412c43e820364b3692588f474780491b : FAILURE in 45m 51s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/4d93828765b243cf83fc0f44f7f1c211 : FAILURE in 1h 00m 30s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/050b481e75ba47f4b5b0df124d6a52a8 : FAILURE in 38m 20s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/6a6fbf809aa04655821eecdf7d00b0d7 : FAILURE in 49m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86f6fa47c39a07d2318a4d25b034b9e9cc23929e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-09-26 07:25:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9497f2304ac7f79c2c125e23005b0878b7a7deed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 08:34:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f61274dd890a40fb98dfe3bb528d358a : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaf97e1ff2c848b187680ddb1c33cc38 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21394e36b1f44198a5428f8f4524a0f1 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaaa17a7bbe1464bb018953128f178c5 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b3c5312eb414d1180f9cb7696691a23 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84fd11b641bf4c40ba369738f7e7123f : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e87fcfe6a34b95b0cafddaf5bc911f : SUCCESS in 7m 25s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d92eeeeb647b4005bb624d9142fb5120 : FAILURE in 34m 37s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/f4b89e38b70249b0aebef6340846c92c : FAILURE in 42m 46s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b00a960335064c85801a78f3d6c593da : FAILURE in 1h 02m 34s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/fb214596f27d42e8aed6d65f7ba3564b : SUCCESS in 35m 25s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/a4aca1281fab4afd8048666b4b3b3fce : FAILURE in 40m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"824d4cca78a6f3a7f27a7e28a2293504d02be48d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-10-01 14:21:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa0ea7182e843ee66154404fede80c49017ac9d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 15:32:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/333f623addf14f348b7e947943e17e74 : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e37b171ba2be4fd2a630d037aa2f60f1 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af8e50712419471fb5dcff69700a6cef : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c74e275b4cbc4bbc962063c7c8b8aa34 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4240ad428094278a2fc99806c6f816e : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ffed2e3b2c645b5b6c1ea6a2c830860 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52b0de04863b4d9db16b2c644f368d95 : SUCCESS in 6m 53s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d2e9a23536184fa080622bec7d927a38 : SUCCESS in 42m 34s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e414d28b7c6c40708f70d288718d3f8b : SUCCESS in 43m 32s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/057889e1be0d41e69155f78f0046624a : SUCCESS in 1h 08m 41s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/cae5a1bed87744a1a3ec6c74d43b7559 : FAILURE in 38m 56s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/c490c7141088479888f3fb2b10826951 : SUCCESS in 40m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57a247389a8b662644ddccac4628b7d68d692884","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-10-04 04:17:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca5cde2477e52e3fdc02d812947c9edf80d9c9c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 04:18:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e47ae75555bfa49a78fe88bdc14ec268cb948112","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 05:22:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bcc13ce883e4b5fa4c17e0f9d1c086a : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c65512d79f84817a1a820ee5861d6c0 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d2f5485bd8f403a815616d36e07c1ed : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e416e56711ed4d6b9f2e5661d5f73b98 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59809802dfd438399d46e8d96d04b4c : SUCCESS in 6m 20s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/3e292ce2abe2429b8b9cec0352dd9231 : SUCCESS in 49m 13s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/85c35a5547c642efa160f682ad246851 : SUCCESS in 1h 00m 29s","accounts_in_message":[],"_revision_number":1},{"id":"85caf3a7a5f5909e03e964a03635fb73a9ebded4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 05:22:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5f0a4c60709daa5c2ff634db7dd769ad28385c50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 05:26:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e07076b60564f538a8f45e39a607408 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23d2e3f02ae960cc9970bc2ef576a7c875cde4d7","revisions":{"23d2e3f02ae960cc9970bc2ef576a7c875cde4d7":{"kind":"REWORK","_number":1,"created":"2021-09-15 02:11:36.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/82/809082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/82/809082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/809082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/809082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/82/809082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/82/809082/1"}}},"commit":{"parents":[{"commit":"ca703064cd24754a7165ffda94b54e08ff6a5525","subject":"Update master for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ca703064cd24754a7165ffda94b54e08ff6a5525"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-15 02:11:33.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-15 02:11:33.000000000","tz":0},"subject":"Add Python3 yoga unit tests","message":"Add Python3 yoga unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for yoga.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I86bcd30139720be4c5949a69075efefaa737990c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/23d2e3f02ae960cc9970bc2ef576a7c875cde4d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/23d2e3f02ae960cc9970bc2ef576a7c875cde4d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
