)]}'
{"id":"openstack%2Fzun~818072","triplet_id":"openstack%2Fzun~master~Ie034819ac90c0fca8c0432cf6c538642ddb45c28","project":"openstack/zun","branch":"master","topic":"ycj","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2021-11-28 07:05:29.000000000","reason":"hongbin replied on the change"},"34189":{"account":{"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"},"last_update":"2021-11-30 01:00:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie034819ac90c0fca8c0432cf6c538642ddb45c28","subject":"Increase the weight of the scheduling strategy.","status":"ABANDONED","created":"2021-11-16 11:54:28.000000000","updated":"2021-11-30 01:00:11.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b10682e6cb0be2edce6ce4ba7b9697c2250d7122","_number":818072,"virtual_id_number":818072,"owner":{"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-16 13:07:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-17 01:45:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-16 13:07:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-17 01:45:56.000000000","updated_by":{"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"CC"},{"updated":"2021-11-18 01:45:42.000000000","updated_by":{"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"dba6ed3ba22cafc32d7c1fe7392d2d9d615e9ae7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"},"date":"2021-11-16 11:54:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36afe296485ce26eb9447236d80a63b3d1a1784a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 13:07:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e0d3ed9085847718c79f237862b2fe4 : SUCCESS in 8m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee9bce7c89f54dd59d4165e7ff067c5c : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18ee38cf7faa433b99dd58c03b44aa41 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0e592b7c3ee4992af6eac43969b6200 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf313c6e08bc4b83949da330e99538a2 : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ff3940ab621459ea02887ac2a9783e4 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4e8d8815ca44fe9694fce40f58b1db : SUCCESS in 7m 29s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/096ab201fdfb425f8bc6c4d1cdd5b956 : SUCCESS in 42m 58s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/6dd6c8faf3ba473a968a6cb97a48ab8f : FAILURE in 22m 04s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/7e55cd1060ce48a69f14e8272bc57caf : SUCCESS in 1h 04m 48s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/d6b3204ba48b49029b597794a699db8c : FAILURE in 32m 00s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/cf795406524c47e09d9287ddfafd65ec : SUCCESS in 45m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97fd26152561b56f9f347293c6140bb5fc3f5033","author":{"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"},"date":"2021-11-17 01:45:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e86ad5ea1c7be98c087405054ccb231eae7c76fe","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-11-28 07:05:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b10682e6cb0be2edce6ce4ba7b9697c2250d7122","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"},"date":"2021-11-30 01:00:11.000000000","message":"Abandoned\n\nclean code.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ef1d26d687dc0609e9a0747f76a08734a368221","revisions":{"7ef1d26d687dc0609e9a0747f76a08734a368221":{"kind":"REWORK","_number":1,"created":"2021-11-16 11:54:28.000000000","uploader":{"_account_id":34189,"name":"yangchangjia","display_name":"Frank","email":"ycj52011@gmail.com","username":"ycj"},"ref":"refs/changes/72/818072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/72/818072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/72/818072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/72/818072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/72/818072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/72/818072/1"}}},"commit":{"parents":[{"commit":"2bf25636eb47c9cc478f463fd4cb96dac436f883","subject":"Increase the weight of the scheduling strategy.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2bf25636eb47c9cc478f463fd4cb96dac436f883"}]}],"author":{"name":"ycj","email":"ycj52011@gmail.com","date":"2021-11-16 11:53:25.000000000","tz":480},"committer":{"name":"ycj","email":"ycj52011@gmail.com","date":"2021-11-16 11:53:34.000000000","tz":480},"subject":"Increase the weight of the scheduling strategy.","message":"Increase the weight of the scheduling strategy.\n\nIn the current version, the node scheduling strategy only filters the resources, and does not calculate the weight of the nodes, resulting in all containers being created on the same node. Similar to nova, it is mainly to increase the weight calculation of the scheduling strategy, which is very necessary. The impact on zun\u0027s overall scheduling process is relatively small. After node filtering, add node weight calculation method, so that zun\u0027s scheduling tasks can be distributed to different nodes, making the scheduling strategy more healthy.\n\nFor the current version, the node hardware configuration of our environment:\n```\ncomputer01\t16H64G7T\tzun-compute\ncomputer02\t16H64G7T\tzun-compute\ncomputer03\t16H64G40T\tnova-compute\ncomputer04\t16H64G7T\tnova-compute\ncontroller01\t16H64G40T\tcontroller\n\n```\nThe \"scheduler\" config of \"zun.conf\" is:\n```\n[scheduler]\ndriver \u003d filter_scheduler\n\navailable_filters \u003d zun.scheduler.filters.all_filters\n\nenabled_filters \u003d CPUFilter\n```\n\nIn \"zun.scheduler.FilterScheduler._schedule()\", we can get node info:\n```\n[\nComputeNode(architecture\u003d\u0027x86_64\u0027,cpu_used\u003d13.0,cpus\u003d16,created_at\u003d2021-06-29T21:31:15Z,disk_quota_supported\u003dTrue,\ndisk_total\u003d5256,disk_used\u003d130,enable_cpu_pinning\u003dFalse,hostname\u003d\u0027computer02\u0027,kernel_version\u003d\u00273.10.0-862.el7.x86_64\u0027,\nlabels\u003d{},mem_available\u003d59848,mem_free\u003d57234,mem_total\u003d63890,mem_used\u003d6656,numa_topology\u003dNUMATopology,\nos\u003d\u0027CentOS Linux 7 (Core)\u0027,os_type\u003d\u0027linux\u0027,paused_containers\u003d0,pci_device_pools\u003dPciDevicePoolList,\nrp_uuid\u003d28f40aa4-aec9-4225-b1d3-b040c8adf24e,running_containers\u003d13,\nruntimes\u003d[\u0027io.containerd.runc.v2\u0027,\u0027io.containerd.runtime.v1.linux\u0027,\u0027runc\u0027],\nstopped_containers\u003d13,total_containers\u003d14,updated_at\u003d2021-11-08T03:47:16Z,uuid\u003d28f40aa4-aec9-4225-b1d3-b040c8adf24e),\nComputeNode(architecture\u003d\u0027x86_64\u0027,cpu_used\u003d1.0,cpus\u003d16,created_at\u003d2021-06-29T21:31:08Z,disk_quota_supported\u003dTrue,\ndisk_total\u003d5356,disk_used\u003d10,enable_cpu_pinning\u003dFalse,hostname\u003d\u0027computer01\u0027,kernel_version\u003d\u00273.10.0-862.el7.x86_64\u0027,\nlabels\u003d{},mem_available\u003d61946,mem_free\u003d63378,mem_total\u003d63890,mem_used\u003d512,numa_topology\u003dNUMATopology,\nos\u003d\u0027CentOS Linux 7 (Core)\u0027,os_type\u003d\u0027linux\u0027,paused_containers\u003d0,pci_device_pools\u003dPciDevicePoolList,\nrp_uuid\u003daa5355b2-5ce8-4108-95af-5d724c2d6dc3,running_containers\u003d1,\nruntimes\u003d[\u0027io.containerd.runc.v2\u0027,\u0027io.containerd.runtime.v1.linux\u0027,\u0027runc\u0027],\nstopped_containers\u003d1,total_containers\u003d1,updated_at\u003d2021-11-08T03:47:14Z,uuid\u003daa5355b2-5ce8-4108-95af-5d724c2d6dc3)\n]\n```\n\nBut the final selected node is:\n```\ncomputer02 ram: 57234MB disk: 5126GB cpus: 3.0\n```\n\nFinally, the \"scheduler\" sample config of \"zun.conf\" is:\n```\n[scheduler]\ndriver \u003d filter_scheduler\n\navailable_filters \u003d zun.scheduler.filters.all_filters\n\nenabled_filters \u003d CPUFilter\n\nweight_classes\u003dzun.scheduler.weights.all_weighers\n\n```\n\nChange-Id: Ie034819ac90c0fca8c0432cf6c538642ddb45c28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7ef1d26d687dc0609e9a0747f76a08734a368221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7ef1d26d687dc0609e9a0747f76a08734a368221"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
