)]}'
{"id":"openstack%2Fzun~832536","triplet_id":"openstack%2Fzun~master~I45ed125d09960c996c7eb5f905ceee5275ea8b88","project":"openstack/zun","branch":"master","topic":"add-zed-python-jobtemplates","hashtags":[],"change_id":"I45ed125d09960c996c7eb5f905ceee5275ea8b88","subject":"Add Python3 zed unit tests","status":"MERGED","created":"2022-03-08 11:43:23.000000000","updated":"2022-04-26 10:16:02.000000000","submitted":"2022-04-26 10:12:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832536-add-zed-python-jobtemplates","meta_rev_id":"6b912a38d03671777e04e89e22e8c11b411a8d10","_number":832536,"virtual_id_number":832536,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-26 10:12:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2022-04-26 08:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2022-04-26 08:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-08 14:07:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 12:29:57.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"CC"},{"updated":"2022-04-26 08:59:24.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"d13940d5bf18f49f21b82783491442656da69a13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-03-08 11:43:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f7e6e12a452bd511693156aed59cef357c6b4f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 14:07:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec57f85df321471199e360745eabedb6 : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/854d5ca1229a4f2399ed6807315a6df9 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d69e4f732a4649889068fcb3eb6f8d16 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef100a3cc17247e7b533e46d32a647a4 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09bced02e28a4fe5be94803156ea3b25 : SUCCESS in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a48df2e434194a3e94addb4937484f57 : RETRY_LIMIT in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26495ae7643442e29158e3bb25f15acf : SUCCESS in 7m 23s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/aecf718361ab469f90b1c15f0f7f72df : TIMED_OUT in 1h 15m 20s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/cbbc5d1a16ae44419ccba0bad367093b : TIMED_OUT in 1h 15m 25s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/74c2a0a25e454ba897915bcdfe3cb4ee : TIMED_OUT in 2h 15m 34s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/f716430eafe6400fa8947fb170cd5648 : FAILURE in 44m 15s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/a593195af85848eda58bd8b9643a3f36 : TIMED_OUT in 1h 15m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aabd0f8d78cdd25a9a1bf5ff2e984545af25f9b1","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-03-10 12:29:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a40f07845969626e3ab2b4c885e720ca92f28fca","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-04-26 06:57:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccc8cdcd14e7e6c6b2b8469a771955293c82a60c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 08:02:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd5eb028a8e248e29ed0dc7dcc7c9c87 : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9286d13254844cb89ee995eb5434fee3 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e26d5105d3d74bb9af8cae88bea192b2 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/486a073e13d74ccaa687810b9f8cdef3 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ec858f3a0154f048d7b37c2c5cc5c0a : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87a7396aee8340d9a09fdc72ba34bc51 : RETRY_LIMIT in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7aaee2728354c33a71887c3592a842a : SUCCESS in 7m 51s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/17ef11f4aef5449ba937595a6c279916 : SUCCESS in 38m 33s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/1611751bc4354b98acbca5e1542285a0 : SUCCESS in 39m 06s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/c6fb892be9d34a67a02a439b3f0e2e33 : SUCCESS in 1h 00m 25s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/4362940ade6b4884b147ce80b9198ffc : FAILURE in 28m 39s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/68e182c1793b459588fa6bcf72fd9915 : SUCCESS in 36m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae69a4e344202892081acd4f23aea7aef20bcfb6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-04-26 08:59:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bdf54595a0724526a048fb9151602510e4115a04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 09:00:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f95bd468cd4ca5125941fd0b7c162dab4210bc90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 10:12:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2197d2055c594dd68ff6d379b430eb6b : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dfc44cb65094af2b8a81c2cbaf312ae : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e95f8efcdaa47a7bff50ef9841be05d : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/febebeeffc4a4383b52df70151e19408 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac5fbe17ed14106bb6298918b314a6e : SUCCESS in 7m 16s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/c44c755aa7434c90aa8bec77fb74523b : SUCCESS in 53m 42s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/334732b5b8214d749336816f1efbe42d : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":1},{"id":"bbf4431f3afeb38ae5862afb49254cba9be013cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 10:12:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6b912a38d03671777e04e89e22e8c11b411a8d10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 10:16:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2624ec57b3742e99119aae4f3d100aa : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"948f69401a504207f65e5a7ce1b2432185d8162c","revisions":{"948f69401a504207f65e5a7ce1b2432185d8162c":{"kind":"REWORK","_number":1,"created":"2022-03-08 11:43:23.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/36/832536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/36/832536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/36/832536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/36/832536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/36/832536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/36/832536/1"}}},"commit":{"parents":[{"commit":"2e8825a05f397b01b760670d606f3c3cb2a8d3cc","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2e8825a05f397b01b760670d606f3c3cb2a8d3cc"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-08 11:43:20.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-08 11:43:20.000000000","tz":0},"subject":"Add Python3 zed unit tests","message":"Add Python3 zed unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for zed.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I45ed125d09960c996c7eb5f905ceee5275ea8b88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/948f69401a504207f65e5a7ce1b2432185d8162c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/948f69401a504207f65e5a7ce1b2432185d8162c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
