)]}'
{"id":"openstack%2Fzun~841422","triplet_id":"openstack%2Fzun~stable%2Fyoga~I04bf9d5dc0955abe832586070e92b3c6c7f76eb0","project":"openstack/zun","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2022-05-12 02:28:08.000000000","reason":"hongbin replied on the change"}},"hashtags":[],"change_id":"I04bf9d5dc0955abe832586070e92b3c6c7f76eb0","subject":"Drop lower constraint job","status":"MERGED","created":"2022-05-11 12:52:12.000000000","updated":"2022-05-23 03:41:25.000000000","submitted":"2022-05-23 03:39:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841422","meta_rev_id":"a034a042d94062a08cf22b204d9debc1e75a7dc1","_number":841422,"virtual_id_number":841422,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-23 03:39:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2022-05-23 02:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2022-05-23 02:48:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-11 15:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0291cc7a8aaf840169436f5f730ddb4a33dcaaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-05-11 12:52:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"290bddccf080c1dd110ddee747a3d80f38b2aa28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 15:12:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bcab4b86b35e4fd992c7286d1e3e2078 : SUCCESS in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/041cfd178f294caca5366157a2e935cc : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9726201cc1204807ba2418c1d68bb2a7 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e52bc26910cb44fe9c333d09220ac403 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18e50bf3e1bf4f288ae8798f3b7d0e20 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d860959a6240c985adac4e0cd436ff : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1b529c11f294f04862309335ffb0be1 : SUCCESS in 3m 35s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7398727166534167aa7836102cde31b2 : TIMED_OUT in 1h 13m 08s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/8ae51f066e6943f3a7fcbbf73eb724d7 : TIMED_OUT in 1h 13m 10s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/05a32a14b0bc48d78d7227f6c5030498 : TIMED_OUT in 2h 13m 58s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/24577835bf3c43e7b19f1d2ecadd5ad2 : FAILURE in 48m 11s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/ea1fa7750c724fb8b9493790e374dd98 : TIMED_OUT in 1h 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60afccbd449c789709cf0b809342bbf6ab35ee9b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-05-12 02:28:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"804a3616d2f1660701522ae561c07e3548e498ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 03:50:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/baca597df20c46e4887ef861567eda98 : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21558dc6a39b4c4e90c991ea4703ad43 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e76d089d4e11441fad5d5471b23e8e65 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29f33e5d6d2d43f3a25cc0bcccb72aac : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe192a8ac6194ef18711524e9ea4dc3a : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/910492cecef7486e95eae6cafe2fe873 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40e5dc53092a44f9838ff831e9751f3a : SUCCESS in 3m 55s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a230d330e2c14845b48657caa221e977 : SUCCESS in 52m 57s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e7ef1ecf729d4ed693ace5d5eb3e7aae : SUCCESS in 50m 18s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/01e72333d9114507b6732f55250a95c1 : SUCCESS in 1h 09m 19s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/68aa6eee98144111b08a5d638ab36605 : FAILURE in 27m 18s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/d9fd9612b4e04af5b4388b8f16a15786 : SUCCESS in 41m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f05c767bb907f16ab023960ac0a9cdafee0bc62","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-05-23 02:48:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2aba3d4db875c5b49c4e01c11290bc6ff64e350","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 02:49:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e650c5dc394ea3fd90e75cbd886cd9ab8b8c0cf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 03:39:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c8d49599d3494cfd826649c0e7dc0032 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92a68cfbccea434b8362eb54afa636d0 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78f8b7ae62da41bcae8ee752dff256cd : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cca19bfd4269422fb080612883074d96 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/988066aab2a145b0a99c9188dfacb5f4 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1c8672444334540a203860ff348ba47 : SUCCESS in 5m 26s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/5ee3006f1c3445c8928e228b7eb76ffb : SUCCESS in 42m 15s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/bc6e1eff67664502870fccdc1d80ea23 : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f628e995227b014bf5ad502a70bc2fcf6c75a3d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 03:39:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a034a042d94062a08cf22b204d9debc1e75a7dc1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 03:41:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e96f532a52704dc69d57b0bf5b625274 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/132a50c9b3eb493295afa06bc9f6dda7 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0d0c28dccdfd9ed975ec02cfcdb1f2a7186a516","revisions":{"e0d0c28dccdfd9ed975ec02cfcdb1f2a7186a516":{"kind":"REWORK","_number":1,"created":"2022-05-11 12:52:12.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/841422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/841422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/841422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/841422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/841422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/841422/1"}}},"commit":{"parents":[{"commit":"ae4812fcf2c8805a3ab9643736779b4efba0e0a6","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ae4812fcf2c8805a3ab9643736779b4efba0e0a6"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-05-11 12:34:50.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-05-11 12:52:12.000000000","tz":0},"subject":"Drop lower constraint job","message":"Drop lower constraint job\n\nChange-Id: I04bf9d5dc0955abe832586070e92b3c6c7f76eb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e0d0c28dccdfd9ed975ec02cfcdb1f2a7186a516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e0d0c28dccdfd9ed975ec02cfcdb1f2a7186a516"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
