)]}'
{"id":"openstack%2Fzun~841428","triplet_id":"openstack%2Fzun~stable%2Fvictoria~I04bf9d5dc0955abe832586070e92b3c6c7f76eb0","project":"openstack/zun","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2022-05-11 14:45:50.000000000","reason":"hongbin replied on the change"}},"hashtags":[],"change_id":"I04bf9d5dc0955abe832586070e92b3c6c7f76eb0","subject":"Drop lower constraint job","status":"MERGED","created":"2022-05-11 13:00:43.000000000","updated":"2022-05-11 16:58:50.000000000","submitted":"2022-05-11 16:56:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841428","meta_rev_id":"a9f116ca86df40c1438c046f1a04c7159b4e0883","_number":841428,"virtual_id_number":841428,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 16:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2022-05-11 14:45:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2022-05-11 14:45:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-11 14:25:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a6e8a771e02e7bb2ba3c1f252dce1f8ec8d28d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-05-11 13:00:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6de8fa419f0d86747e68d90ebcf9fb0da34c2c2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-05-11 13:14:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b0875a4a2b65df3ddf5433544aab1e632a05c8fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 14:25:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fb9c15affbb40039d12479c5420fc00 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6d3aa980d964a3eb6b8f1988fc95f42 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b20ef90a67a44a0890d3b0658a7e1497 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96287285ca6a4ff89bce045c4139747b : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b38178720c49c486ddc13e788a165d : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10914d98de9c4053a8a858117002e918 : SUCCESS in 3m 32s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b1ca58f998bd4209bd8ff53a8b51e35c : SUCCESS in 51m 21s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/b5fd01437bf34059a4ba263bf7fe4355 : FAILURE in 56m 29s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/4d061ed22ee740e2b86a00702474d7f4 : FAILURE in 1h 09m 55s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/5e6e4c187d6844bf99adac211c46483b : FAILURE in 32m 40s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/276485878b5d4192a54f5a9cb76e58b6 : FAILURE in 50m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87e69201100b9c1ad174e48b7600df371619e827","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-05-11 14:35:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"902ac960b3ef4de9308823e1f7de03a5739e6dc1","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-05-11 14:45:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cd2e4e0cb87c9c9dfcb464feac58e5af25a747a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 15:48:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2de41a4437246f8acbaeed241f7d318 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6e546beb742478c95ea1de4d91f6fac : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee0e58baa573497e819e1eb50246fbcd : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d68c05a0d6064ed8b6bbee844a1d2c40 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74bab520fc749bdaa2532125ee92555 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abf8c1843ea940c3a65819398bd83b7e : SUCCESS in 3m 43s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b66ae9dabd754bd590705351cd5d17d8 : SUCCESS in 49m 02s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/c574edf1d5ed443a8569e2457e963b31 : SUCCESS in 52m 23s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/cc5883ac9bb240ad885901129962666c : SUCCESS in 1h 06m 17s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/d739b51b1964498b9035a48b910bbe24 : FAILURE in 40m 13s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/b1b9ee1d9e8f4c9fab70b150f47b2882 : SUCCESS in 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6c6532613551c6cdf36e165cce6e7e17511d04fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 15:50:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fa1c8f17f84f2df4ab43f5831f26889ae8da3c16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:56:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88bfa42f9f00451187ab1e40324d22a4 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2131688c594046f38f99537f4830eabd : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbbcaf6f027048b398747140b891020a : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2875fff7897e4af19d123142ca9783fa : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86264e46a9534d3c82bced2b2f4c5f23 : SUCCESS in 4m 35s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9a60325bce41452388afc6b96d47e123 : SUCCESS in 1h 00m 37s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/eac77e99bdf84ea797ba9a54bd0a89a5 : SUCCESS in 45m 13s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/baf7eb805f334f119e55ac0803393395 : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":3},{"id":"253934fb9214a659108450436efa435415e58c84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:56:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a9f116ca86df40c1438c046f1a04c7159b4e0883","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:58:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e75d6f5355e4998bb81e3d834033163 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5df1bf2ac8684aaeb99cb804f31c5f32 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"00b8d89e0442a892bc588c3224608c5df680875a","revisions":{"720eb05532f59a9e4d08ef685477aeee0206206a":{"kind":"REWORK","_number":1,"created":"2022-05-11 13:00:43.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/841428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/841428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/841428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/841428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/841428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/841428/1"}}},"commit":{"parents":[{"commit":"4014ba9ea326fc15f370c2dda5cecf046275e429","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4014ba9ea326fc15f370c2dda5cecf046275e429"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-05-11 12:34:50.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-05-11 13:00:42.000000000","tz":0},"subject":"Drop lower constraint job","message":"Drop lower constraint job\n\nChange-Id: I04bf9d5dc0955abe832586070e92b3c6c7f76eb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/720eb05532f59a9e4d08ef685477aeee0206206a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/720eb05532f59a9e4d08ef685477aeee0206206a"}]},"branch":"refs/heads/stable/victoria"},"c0b9266587be83f30a9ed378ddb05a6825f9d3de":{"kind":"REWORK","_number":2,"created":"2022-05-11 13:14:12.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/841428/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/841428/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/841428/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/841428/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/841428/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/841428/2"}}},"commit":{"parents":[{"commit":"4014ba9ea326fc15f370c2dda5cecf046275e429","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4014ba9ea326fc15f370c2dda5cecf046275e429"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-05-11 12:34:50.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-05-11 13:14:06.000000000","tz":0},"subject":"Drop lower constraint job","message":"Drop lower constraint job\n\nChange-Id: I04bf9d5dc0955abe832586070e92b3c6c7f76eb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c0b9266587be83f30a9ed378ddb05a6825f9d3de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c0b9266587be83f30a9ed378ddb05a6825f9d3de"}]},"branch":"refs/heads/stable/victoria"},"00b8d89e0442a892bc588c3224608c5df680875a":{"kind":"REWORK","_number":3,"created":"2022-05-11 14:35:32.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/841428/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/28/841428/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/841428/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/841428/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/28/841428/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/28/841428/3"}}},"commit":{"parents":[{"commit":"4014ba9ea326fc15f370c2dda5cecf046275e429","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4014ba9ea326fc15f370c2dda5cecf046275e429"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-05-11 12:34:50.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-05-11 14:35:26.000000000","tz":0},"subject":"Drop lower constraint job","message":"Drop lower constraint job\n\nChange-Id: I04bf9d5dc0955abe832586070e92b3c6c7f76eb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/00b8d89e0442a892bc588c3224608c5df680875a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/00b8d89e0442a892bc588c3224608c5df680875a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
